)]}'
{"id":"openstack%2Fmulti-arch-sig~733192","triplet_id":"openstack%2Fmulti-arch-sig~master~I65c304c9b6edccac620118ae76fc55eebcc59fc8","project":"openstack/multi-arch-sig","branch":"master","hashtags":[],"change_id":"I65c304c9b6edccac620118ae76fc55eebcc59fc8","subject":"Make some simple doc enhancements","status":"MERGED","created":"2020-06-03 16:01:31.000000000","updated":"2020-08-04 15:19:29.000000000","submitted":"2020-08-04 15:17:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733192-1596554232783-60d190a1","meta_rev_id":"dc30825e62d7f8b40072262ff093c14fd17a67fe","_number":733192,"virtual_id_number":733192,"owner":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 15:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2020-08-04 15:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2020-08-04 15:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 16:02:03.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-06-03 16:02:03.000000000","updated_by":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"reviewer":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"state":"REVIEWER"},{"updated":"2020-06-03 16:02:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-06-03 16:02:03.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-06-04 06:16:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-04 15:04:44.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-08-04 15:17:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a422780e7d396c2d518ff6d2c0cd8bdc31b8df63","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-06-03 16:01:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbe4b71b9a9c76dc043e4283516944cada113454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:06:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7958a60dfc4a609970f8ef24af0c0c : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bcc09c723b63fa46672e87b1c6a34778b42b5834","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-04 06:16:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51efdffd4a27e016d62b3b2feb9e9aa6844b4e6a","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-08-04 15:04:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3e57081a8d0436fa4136ec281e7ce311604bca63","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-08-04 15:04:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"12904ab3f32feb549529a24dfb49688378062db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:10:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7996f8b14f94a54b6e130e04479ce58 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b551192b99926ad70b95ca3464c76afc500a2c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:10:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7e44bdc59a1efbbe543597733d9506af752b214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:17:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306df16214434de1807389a4748288b7 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0d36fb4883485c4f902367c96907087e6f5e2f63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:17:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dc30825e62d7f8b40072262ff093c14fd17a67fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:19:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5de84998c347e7b8bca1234a417622 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b32ac2d153585c252f66c4b3784492fb7a450b76","revisions":{"92fb810d63f001b01b195137d84facb7b6bdb848":{"kind":"REWORK","_number":1,"created":"2020-06-03 16:01:31.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/92/733192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/92/733192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/1"}}},"commit":{"parents":[{"commit":"cda23cdc8da1796b761173278a99d4bec7137a64","subject":"Preliminary docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/cda23cdc8da1796b761173278a99d4bec7137a64"}]}],"author":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-06-03 16:00:47.000000000","tz":-240},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-06-03 16:00:47.000000000","tz":-240},"subject":"Make some simple doc enhancements","message":"Make some simple doc enhancements\n\nBasic enhancements\nPlease note that the goal is to link from this page to a more specific\nstatus for each of the various aspects.\n\nChange-Id: I65c304c9b6edccac620118ae76fc55eebcc59fc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/92fb810d63f001b01b195137d84facb7b6bdb848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/92fb810d63f001b01b195137d84facb7b6bdb848"}]},"branch":"refs/heads/master"},"b32ac2d153585c252f66c4b3784492fb7a450b76":{"kind":"REWORK","_number":2,"created":"2020-08-04 15:04:22.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/92/733192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/92/733192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/92/733192/2"}}},"commit":{"parents":[{"commit":"cda23cdc8da1796b761173278a99d4bec7137a64","subject":"Preliminary docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/cda23cdc8da1796b761173278a99d4bec7137a64"}]}],"author":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-06-03 16:00:47.000000000","tz":-240},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-08-04 15:04:19.000000000","tz":0},"subject":"Make some simple doc enhancements","message":"Make some simple doc enhancements\n\nBasic enhancements\nPlease note that the goal is to link from this page to a more specific\nstatus for each of the various aspects.\n\nChange-Id: I65c304c9b6edccac620118ae76fc55eebcc59fc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/b32ac2d153585c252f66c4b3784492fb7a450b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/b32ac2d153585c252f66c4b3784492fb7a450b76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
