)]}'
{"id":"openstack%2Fmulti-arch-sig~790595","triplet_id":"openstack%2Fmulti-arch-sig~master~Ie2042479be6bd022391c610a2dfe9f42fa526e59","project":"openstack/multi-arch-sig","branch":"master","hashtags":[],"change_id":"Ie2042479be6bd022391c610a2dfe9f42fa526e59","subject":"setup.cfg: Replace dashes with underscores","status":"ABANDONED","created":"2021-05-11 08:53:50.000000000","updated":"2021-12-23 08:19:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"369ff6b31043b122b25fa0ce2859544d2de695cb","_number":790595,"virtual_id_number":790595,"owner":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-11 09:03:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 09:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 08:17:36.000000000","updated_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"30b52eeab99a788e4010336b1d52614e90624162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-05-11 08:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6932abca16c6db6f3907ff6cb32261096e138905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ff531d9a4d4417b810d59185323833 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"369ff6b31043b122b25fa0ce2859544d2de695cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-12-23 08:19:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7aa7a1f07a3e21d44314fa7571bc8bcbb9217f7b","revisions":{"7aa7a1f07a3e21d44314fa7571bc8bcbb9217f7b":{"kind":"REWORK","_number":1,"created":"2021-05-11 08:53:50.000000000","uploader":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"ref":"refs/changes/95/790595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/95/790595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/95/790595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/95/790595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/95/790595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/95/790595/1"}}},"commit":{"parents":[{"commit":"50655134172a4bdac07b76103eead563678f8101","subject":"Merge \"Make some simple doc enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/50655134172a4bdac07b76103eead563678f8101"}]}],"author":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-05-11 08:52:49.000000000","tz":480},"committer":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-05-11 08:53:46.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated options in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without this, we see \u0027UserWarning\u0027 messages like the following on new enough versions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ie2042479be6bd022391c610a2dfe9f42fa526e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/7aa7a1f07a3e21d44314fa7571bc8bcbb9217f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/7aa7a1f07a3e21d44314fa7571bc8bcbb9217f7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
