)]}'
{"id":"openstack%2Fmurano-agent~665879","triplet_id":"openstack%2Fmurano-agent~master~I698d48349b94ccf8fb87e97e9681a5412336eb78","project":"openstack/murano-agent","branch":"master","hashtags":[],"change_id":"I698d48349b94ccf8fb87e97e9681a5412336eb78","subject":"Update contraints url","status":"MERGED","created":"2019-06-18 06:39:44.000000000","updated":"2019-06-18 07:24:55.000000000","submitted":"2019-06-18 07:24:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665879-1560842695596-1e9fde82","meta_rev_id":"443c7d769b3fdf53ccd029b8f4b11a4007c707f3","_number":665879,"virtual_id_number":665879,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-18 07:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-06-18 07:06:06.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-06-18 07:06:06.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 07:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd8ec8fbd05279975c7f14ea5855ea15c7441364","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 06:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0057ae4b08d5af71ebc1d0960512079f9db23b87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 06:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/79/665879/1/check/openstack-tox-lower-constraints/d5d3d4c/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/79/665879/1/check/openstack-tox-pep8/56dc369/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/79/665879/1/check/openstack-tox-py27/a9bf044/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/79/665879/1/check/openstack-tox-py36/39c2aa4/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/79/665879/1/check/openstack-tox-py37/1b76624/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/79/665879/1/check/build-openstack-releasenotes/170af7f/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7e331b0769d6d4b1afc79248da8660070d4b1524","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 07:06:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a07f81487873d7f75c29edaf5f546f027be44757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:06:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5d83e5cf975b500778c67d5f08d47062a7ed06e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:24:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"443c7d769b3fdf53ccd029b8f4b11a4007c707f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:24:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/79/665879/1/gate/openstack-tox-lower-constraints/24b3f3f/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/79/665879/1/gate/openstack-tox-pep8/d9c4efb/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/79/665879/1/gate/openstack-tox-py27/588b8ad/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/79/665879/1/gate/openstack-tox-py36/4d728e5/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/79/665879/1/gate/openstack-tox-py37/8855f1d/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/79/665879/1/gate/build-openstack-releasenotes/75a8ff5/html/ : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0161875574c8a55911a8597d6f122301aa0ddfe5","revisions":{"0161875574c8a55911a8597d6f122301aa0ddfe5":{"kind":"REWORK","_number":1,"created":"2019-06-18 06:39:44.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/79/665879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/79/665879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/79/665879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/79/665879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/79/665879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/79/665879/1"}}},"commit":{"parents":[{"commit":"46dadfbedd8f66f82037ddda1ea393de8dff4860","subject":"Fix sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/46dadfbedd8f66f82037ddda1ea393de8dff4860"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:39:22.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:39:22.000000000","tz":480},"subject":"Update contraints url","message":"Update contraints url\n\nChange-Id: I698d48349b94ccf8fb87e97e9681a5412336eb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/0161875574c8a55911a8597d6f122301aa0ddfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/0161875574c8a55911a8597d6f122301aa0ddfe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
