)]}'
{"id":"openstack%2Fmurano-agent~668839","triplet_id":"openstack%2Fmurano-agent~master~I98fd786e292e1da946181f15dd3fe91ff3338508","project":"openstack/murano-agent","branch":"master","hashtags":[],"change_id":"I98fd786e292e1da946181f15dd3fe91ff3338508","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-03 09:21:21.000000000","updated":"2019-07-04 02:48:26.000000000","submitted":"2019-07-04 02:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668839-1562208506482-a663f06a","meta_rev_id":"a66a8792bc0a55f41566b61e75086462cc7750ec","_number":668839,"virtual_id_number":668839,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-07-04 02:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-07-04 02:36:11.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-07-04 02:36:11.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 02:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b05f473e2921fcc1f44927d427bf456d4d4e6b47","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-03 09:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a7866782c791c1cabcdac8d51fea29a2d45740c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:35:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/39/668839/1/check/test-release-openstack/ccedc13/ : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/668839/1/check/openstack-tox-lower-constraints/c5d3115/ : FAILURE in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/39/668839/1/check/openstack-tox-pep8/4c38357/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/39/668839/1/check/openstack-tox-py27/17b19f1/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/39/668839/1/check/openstack-tox-py36/ec9132f/ : SUCCESS in 2m 46s\n- openstack-tox-py37 http://logs.openstack.org/39/668839/1/check/openstack-tox-py37/a33610a/ : SUCCESS in 3m 25s\n- build-openstack-releasenotes http://logs.openstack.org/39/668839/1/check/build-openstack-releasenotes/b324062/html/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"de3ad3884a503b6bfe071f32b219970d93190a7e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 00:53:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88f97a629b50cf77ed27ac4ccd7b49761648be40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:09:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/39/668839/1/check/test-release-openstack/4cc3eef/ : SUCCESS in 1m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/668839/1/check/openstack-tox-lower-constraints/a93b84e/ : SUCCESS in 2m 44s\n- openstack-tox-pep8 http://logs.openstack.org/39/668839/1/check/openstack-tox-pep8/9fb4f1c/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/39/668839/1/check/openstack-tox-py27/548f899/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/39/668839/1/check/openstack-tox-py36/7603ed3/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/39/668839/1/check/openstack-tox-py37/7ba69b8/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes http://logs.openstack.org/39/668839/1/check/build-openstack-releasenotes/069e60e/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0880505e9ce5f4d8d9b765d5dc7179729572df35","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 02:36:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a7f946e3ac87537e9f0188ed05d5710b8f4c2e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:36:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48a184e7d296a827c880417cdd63cb4f24142e82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:48:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/39/668839/1/gate/test-release-openstack/a8a6992/ : SUCCESS in 2m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/668839/1/gate/openstack-tox-lower-constraints/ee5c1f0/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/39/668839/1/gate/openstack-tox-pep8/cbd7a4f/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/39/668839/1/gate/openstack-tox-py27/9c1174c/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/39/668839/1/gate/openstack-tox-py36/334e922/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/39/668839/1/gate/openstack-tox-py37/698624e/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/39/668839/1/gate/build-openstack-releasenotes/b9c3b15/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a66a8792bc0a55f41566b61e75086462cc7750ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:48:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a026aa2e2f38bcfe072fb9cdfa6e9719f6fda5c6","revisions":{"a026aa2e2f38bcfe072fb9cdfa6e9719f6fda5c6":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:21:21.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/39/668839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/39/668839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/39/668839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/39/668839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/39/668839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/39/668839/1"}}},"commit":{"parents":[{"commit":"0161875574c8a55911a8597d6f122301aa0ddfe5","subject":"Update contraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/0161875574c8a55911a8597d6f122301aa0ddfe5"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-03 09:20:47.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-03 09:20:47.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I98fd786e292e1da946181f15dd3fe91ff3338508\nStory: #2005924\nTask: #34214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/a026aa2e2f38bcfe072fb9cdfa6e9719f6fda5c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/a026aa2e2f38bcfe072fb9cdfa6e9719f6fda5c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
