)]}'
{"id":"openstack%2Fmurano-agent~725671","triplet_id":"openstack%2Fmurano-agent~master~I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38","project":"openstack/murano-agent","branch":"master","topic":"bug/1863021","hashtags":[],"change_id":"I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-05 19:41:19.000000000","updated":"2020-05-06 08:44:27.000000000","submitted":"2020-05-05 21:29:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725671-1588714182728-3c423189","meta_rev_id":"4beefdd741e93fb5b67965991fe462f0de9ff46d","_number":725671,"virtual_id_number":725671,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 21:29:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},"all":[{"value":2,"date":"2020-05-05 21:20:25.000000000","_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},"all":[{"value":1,"date":"2020-05-05 21:20:25.000000000","_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 21:20:25.000000000","updated_by":{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},"reviewer":{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},"state":"REVIEWER"},{"updated":"2020-05-05 21:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f5010c598c4781a40662862a6ec2d7767397782","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-05 19:41:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af504dbbf4539e2850e908d3fb849f94017681ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 19:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fe6292b0c3d48d1aa7c59398e8c24ff : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c840a56331a4b988a0a3bb59b3c54e2 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567985af7dd54d83a80b66abaa56aa2c : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54efd86e857c4db5adf20cfc6449d96c : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ee3d3ac1f32e2d71892e5ca2e8298aaa8eb43ee4","author":{"_account_id":15168,"name":"Kirill Zaitsev","email":"k.zaitsev@me.com","username":"kzaitsev"},"date":"2020-05-05 21:20:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c61e2a594c26ba56c3f225fd69e467747b53551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 21:20:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f3588e9cccc651da15cdb2aba50fd4f05168e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 21:29:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36abf5967c2e4fd5b939f6b9c2042a08 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a778205b46f4c3dad096fbbdf5021a4 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1122b65c25da4a1f9ad87b4fe55f8013 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59fbc41e75884205b3d71a7d391e3c1b : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"04d8de77da62000b7cef1231e45ee5a568484746","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 21:29:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd0429d60ac6bb54823b304042bd02708eb0a05b","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2020-05-06 08:44:27.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 97f53bb54a6209bc917e1a5a727ea8f6333bafc1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2e41f2f659fc4490e55d3372cfbcebb295c8ad4","revisions":{"f2e41f2f659fc4490e55d3372cfbcebb295c8ad4":{"kind":"REWORK","_number":1,"created":"2020-05-05 19:41:19.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/71/725671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/71/725671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/71/725671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/71/725671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/71/725671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/71/725671/1"}}},"commit":{"parents":[{"commit":"eef4d84278af6eae5a700774ae04b470babb37e1","subject":"Merge \"Bump default tox env from py37 to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/eef4d84278af6eae5a700774ae04b470babb37e1"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:41:02.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:41:18.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/f2e41f2f659fc4490e55d3372cfbcebb295c8ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/f2e41f2f659fc4490e55d3372cfbcebb295c8ad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
