)]}'
{"id":"openstack%2Fmurano-agent~725796","triplet_id":"openstack%2Fmurano-agent~stable%2Fussuri~I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38","project":"openstack/murano-agent","branch":"stable/ussuri","topic":"bug/1863021-stable/ussuri","hashtags":[],"change_id":"I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-06 08:44:27.000000000","updated":"2020-05-07 05:57:09.000000000","submitted":"2020-05-07 05:57:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725796-1588831029752-0280b767","meta_rev_id":"96dff9b5d82eb69763d3dff056be2553d1b2c8ff","_number":725796,"virtual_id_number":725796,"owner":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 05:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-05-07 05:49:42.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-05-07 05:49:42.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-06 08:44:27.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 08:44:27.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2020-05-07 05:49:42.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-05-07 05:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dea8c40477ad083fc9c31baad4db8776cf60ccd9","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2020-05-06 08:44:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ab221aa5dab13cacbeaf50c932c6adff799dec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 08:53:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22d26d945c07483fad5d46d03690609b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548ea32227ca4be5bb3fda3d7d5c3519 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd1526679af45f8891e5ed880924c45 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b27746caf9b547e7a8ea1c755b663cd6 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"08513272126c25df80c0fc1e5da18a2c43b445ef","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-05-07 05:49:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7182567b2173bb132cdcd118dd3b924fa536c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 05:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d343263878c5e738dc49487f9d3a2dc1f8925f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 05:57:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1d07894fa84f668eecec38ec91a7d7 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beed15bdf5454fd691328f3fec3100b7 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91e3f5fdefd2471086e82e3deb5c04fb : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/278c446b249a41909f5d8b55182fa8d4 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d1f9b1d3acbeade6d0e84fc37b8b675e02f65913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 05:57:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97f53bb54a6209bc917e1a5a727ea8f6333bafc1","revisions":{"97f53bb54a6209bc917e1a5a727ea8f6333bafc1":{"kind":"REWORK","_number":1,"created":"2020-05-06 08:44:27.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/96/725796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/96/725796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/96/725796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/96/725796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/96/725796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/96/725796/1"}}},"commit":{"parents":[{"commit":"2b16c3beb52789daf062f64b633446284a19a078","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/2b16c3beb52789daf062f64b633446284a19a078"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-05 19:41:02.000000000","tz":-240},"committer":{"name":"Chris MacNaughton (icey)","email":"chris.macnaughton@canonical.com","date":"2020-05-06 08:44:27.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I0e8f80bb2e7acfb00e5ff47e0aa983afb7ee6e38\nCloses-Bug: #1863021\n(cherry picked from commit f2e41f2f659fc4490e55d3372cfbcebb295c8ad4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/97f53bb54a6209bc917e1a5a727ea8f6333bafc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/97f53bb54a6209bc917e1a5a727ea8f6333bafc1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
