)]}'
{"id":"openstack%2Fmurano-agent~861019","triplet_id":"openstack%2Fmurano-agent~master~I00ae5e9545136c1b930f8cf4036b902a326e1c54","project":"openstack/murano-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"18418":{"account":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"last_update":"2022-10-14 01:38:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00ae5e9545136c1b930f8cf4036b902a326e1c54","subject":"Remove use of anyjson module","status":"MERGED","created":"2022-10-12 02:55:48.000000000","updated":"2022-10-14 01:38:12.000000000","submitted":"2022-10-14 01:38:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861019","meta_rev_id":"f0a158febadd1423e7e5d4acc7752898b95409cd","_number":861019,"virtual_id_number":861019,"owner":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 01:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2022-10-14 01:28:53.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2022-10-14 01:28:53.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 03:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 01:28:53.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"b945014086629122c63510f5ec8db4a5a91f518c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-12 02:55:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"447ef082436f6f99d1093eab7929441417872cf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e1a77db77af4de481e25e5b7279841c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d41f215fd5f4308b51f5ab6337b50c8 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfd271bb3d64e0b95464f1a3cdaffca : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9e71f58c3ca4e4dba7dc0c4d5cb7592 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b687ac529e734ec6b6b9977fec0e45a3 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dc2a9d6d6094808b6a7541855dc76b7 : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72e6ba16537d90a9ef40b8718eadc9bd870d843e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-10-14 01:28:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8b52309bc9e58c59c42fbbafa6372c9ccfe9509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 01:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13d70eba1c4415d4aa3ee58cb5fb574b435becf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 01:38:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0059eceb13448b0a7d4cc7934876a87\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a65db67e9254a72a368258135c3a00c : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff98fc2db1f47c99bd32c96a8ebbec9 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa7d7c6cb4c4475fbb64f5e3ace4ded2 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed472e8239dc4fd490e77a332fad293e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f0a158febadd1423e7e5d4acc7752898b95409cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 01:38:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"449c50d170fecc814defc39c009a075056e15c58","revisions":{"449c50d170fecc814defc39c009a075056e15c58":{"kind":"REWORK","_number":1,"created":"2022-10-12 02:55:48.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/19/861019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/19/861019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/19/861019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/19/861019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/19/861019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/19/861019/1"}}},"commit":{"parents":[{"commit":"eff24e0861b9aa7b900aeab245030a3e582a0de4","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/eff24e0861b9aa7b900aeab245030a3e582a0de4"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2022-10-12 02:51:37.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2022-10-12 02:55:45.000000000","tz":660},"subject":"Remove use of anyjson module","message":"Remove use of anyjson module\n\nThe use of the anyjson module causes issues with Python 3.10 with\nthe following error:\n`error in anyjson setup command: use_2to3 is invalid.`\n\nThe module itself hasn\u0027t been updated since 2012 and the source [0]\nnow errors with a 404, and there doesn\u0027t seem to be any real benefit\nover the standard python json module.\n\n[0] https://bitbucket.org/runeh/anyjson/\n\nChange-Id: I00ae5e9545136c1b930f8cf4036b902a326e1c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/449c50d170fecc814defc39c009a075056e15c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/449c50d170fecc814defc39c009a075056e15c58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[]}
