)]}'
{"id":"openstack%2Fmurano-agent~895289","triplet_id":"openstack%2Fmurano-agent~stable%2F2023.2~I3aa9d8f767c5ff98dc6cabe47a28296ae652f196","project":"openstack/murano-agent","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"I3aa9d8f767c5ff98dc6cabe47a28296ae652f196","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","status":"MERGED","created":"2023-09-15 13:48:52.000000000","updated":"2023-09-21 15:55:04.000000000","submitted":"2023-09-21 15:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895289-create-2023.2","meta_rev_id":"96999ea4e31764d90cf19649c2d43ecb627e46bb","_number":895289,"virtual_id_number":895289,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 15:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-09-21 15:42:50.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-09-21 15:42:50.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 15:42:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"72a29ee9e4303a1510af426bbaa540d30825ffe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 13:48:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41021d682e967f5a4b22419633224c0c08714621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aae859641ba44aabe143a60988893bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc6f05d9367413a8f711b01321731d6 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a8477d8243e490497c746c4d400d603 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5de75e24255e4ddc9528ed2488631263 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0aa9c96589974d0ca5212b93cace461c : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00f718b702924f97b23702d2372a5fd6 : SUCCESS in 3m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c2832cd577b4ca786e29cbc15713d16 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"80fe11fdaee75141e1d01c630c781152bef1de49","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-21 15:42:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8858856c02cb62fc7d6c6893a3682274e20686d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:43:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff2de05bd223aaeb99de69f0d9cd28f321223a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:53:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a92f75a4db4460392c9636185c4fbe6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e5f0abe405437d97dc8dfa13428b74 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d13b0768b33342f9bc14fe50978aacce : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e63ec1ebcd94a9b8be3639431f508ef : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd24447f775043c090d4174a4ee0b01c : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baed4fb852c546c4bf24c241ff0bf72d : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cb2f325bad45aada87785fa5e8212f71ddb75b99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:53:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"96999ea4e31764d90cf19649c2d43ecb627e46bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:55:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1681a149b1af40778ceb6991844a398a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66869f96b0934d6ca1e62f5d414deeb5 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df378affb8a81f4ffe6d0e33099f5a67d3b17222","revisions":{"df378affb8a81f4ffe6d0e33099f5a67d3b17222":{"kind":"REWORK","_number":1,"created":"2023-09-15 13:48:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/89/895289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-agent","ref":"refs/changes/89/895289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/89/895289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/89/895289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-agent refs/changes/89/895289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-agent refs/changes/89/895289/1"}}},"commit":{"parents":[{"commit":"633801455bd7fc19f7f45e9b472d69cbffd9d312","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/633801455bd7fc19f7f45e9b472d69cbffd9d312"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 13:48:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 13:48:52.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I3aa9d8f767c5ff98dc6cabe47a28296ae652f196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/df378affb8a81f4ffe6d0e33099f5a67d3b17222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-agent/commit/df378affb8a81f4ffe6d0e33099f5a67d3b17222"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
