)]}'
{"id":"openstack%2Fmurano-pkg-check~665883","triplet_id":"openstack%2Fmurano-pkg-check~master~I253de05f298ed33cde6a98455264fb3f7b040388","project":"openstack/murano-pkg-check","branch":"master","hashtags":[],"change_id":"I253de05f298ed33cde6a98455264fb3f7b040388","subject":"Update contraints url","status":"MERGED","created":"2019-06-18 06:50:50.000000000","updated":"2019-06-18 07:18:11.000000000","submitted":"2019-06-18 07:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665883-1560842291515-d473cd44","meta_rev_id":"11c992b1322191679c5bf82ebe62beeca98538b9","_number":665883,"virtual_id_number":665883,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-18 07:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-06-18 07:06:17.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-06-18 07:06:17.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 07:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5eb0dec6bca7b9a3884d8d3d5440281fa4219842","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 06:50:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969360288055d27229ffc4048233a8e292f55b2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/83/665883/1/check/openstack-tox-cover/b9abfbe/cover/ : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/665883/1/check/openstack-tox-lower-constraints/3e7dc1a/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/83/665883/1/check/openstack-tox-pep8/a663ddd/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/83/665883/1/check/openstack-tox-py27/7edb1ac/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/83/665883/1/check/openstack-tox-py36/e7d621d/ : SUCCESS in 6m 45s\n- openstack-tox-py37 http://logs.openstack.org/83/665883/1/check/openstack-tox-py37/c718320/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a90f881d9c1095e2648496c0b3c0d408a081d4f1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 07:06:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbef365510cf4def889ef1360753c3936a2c58d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:06:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00e413d2aaf3396202d230e72db16ea380d829d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:18:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11c992b1322191679c5bf82ebe62beeca98538b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:18:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/665883/1/gate/openstack-tox-lower-constraints/31ab049/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/83/665883/1/gate/openstack-tox-pep8/a984b35/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/83/665883/1/gate/openstack-tox-py27/0650ea8/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/83/665883/1/gate/openstack-tox-py36/e4e20ce/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/83/665883/1/gate/openstack-tox-py37/a779f32/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a78087c189ce25a380b407e804e65d96a45e1a23","revisions":{"a78087c189ce25a380b407e804e65d96a45e1a23":{"kind":"REWORK","_number":1,"created":"2019-06-18 06:50:50.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/83/665883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano-pkg-check","ref":"refs/changes/83/665883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano-pkg-check refs/changes/83/665883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano-pkg-check refs/changes/83/665883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano-pkg-check refs/changes/83/665883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano-pkg-check refs/changes/83/665883/1"}}},"commit":{"parents":[{"commit":"3a4020e35a661a7f6d727281624812501b7e3b52","subject":"Merge \"Fix sphinx requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-pkg-check/commit/3a4020e35a661a7f6d727281624812501b7e3b52"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:50:33.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:50:33.000000000","tz":480},"subject":"Update contraints url","message":"Update contraints url\n\nChange-Id: I253de05f298ed33cde6a98455264fb3f7b040388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-pkg-check/commit/a78087c189ce25a380b407e804e65d96a45e1a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano-pkg-check/commit/a78087c189ce25a380b407e804e65d96a45e1a23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
