)]}'
{"id":"openstack%2Fmurano~654827","triplet_id":"openstack%2Fmurano~master~Id506aad04b878ba516872ef0169664f47f6eaab0","project":"openstack/murano","branch":"master","hashtags":[],"change_id":"Id506aad04b878ba516872ef0169664f47f6eaab0","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 08:15:34.000000000","updated":"2019-06-28 04:25:17.000000000","submitted":"2019-06-28 04:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654827-1561695917833-ca750131","meta_rev_id":"be547dc9fff24b28fb905660672be3ce7e48a841","_number":654827,"virtual_id_number":654827,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-28 04:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-06-28 03:30:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-06-28 03:30:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 03:50:11.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-28 03:30:13.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-06-28 04:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45be1cc83a6333ad8400623b01a6bb044e16b147","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 08:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"738c216ef37a7b0310f023c5d9114adb1320e904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:41:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/27/654827/1/check/test-release-openstack/d627e66/ : SUCCESS in 3m 11s\n- openstack-tox-cover http://logs.openstack.org/27/654827/1/check/openstack-tox-cover/82b4f39/cover/ : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/1/check/openstack-tox-lower-constraints/ecb4013/ : POST_FAILURE in 19m 35s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/1/check/openstack-tox-pep8/8bce956/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/1/check/openstack-tox-py27/8149a98/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/1/check/openstack-tox-py36/bdd67e4/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/1/check/openstack-tox-py37/c13984a/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/27/654827/1/check/openstack-tox-docs/d48ee8c/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/1/check/build-openstack-releasenotes/f452a05/html/ : SUCCESS in 5m 41s\n- murano-congress-devstack http://logs.openstack.org/27/654827/1/check/murano-congress-devstack/c84b24a/ : SUCCESS in 38m 44s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/1/check/murano-rally-task/0f962ff/ : SUCCESS in 41m 05s\n- murano-tempest-api http://logs.openstack.org/27/654827/1/check/murano-tempest-api/c859b5d/ : SUCCESS in 44m 04s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/1/check/murano-tempest-cfapi/4284f9f/ : SUCCESS in 32m 13s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/1/check/murano-grenade/0d61b72/ : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb36ccb296262e758a7d437dbf9acd83976172e2","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-23 11:16:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a943090e4a3e38b534b8e9542cfa9d47bc47dfd","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 08:06:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aec80ce35d115a41654af596100a7f43688cb16","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 08:10:20.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"38dd0cc9e0b218ee1d06172e0a89b04c8b9bab34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:37:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/27/654827/2/check/test-release-openstack/5fe26f8/ : SUCCESS in 3m 18s\n- openstack-tox-cover http://logs.openstack.org/27/654827/2/check/openstack-tox-cover/5efc224/cover/ : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/2/check/openstack-tox-lower-constraints/4be5aaa/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/2/check/openstack-tox-pep8/4999223/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/2/check/openstack-tox-py27/9d6689f/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/2/check/openstack-tox-py36/7806b9a/ : SUCCESS in 6m 42s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/2/check/openstack-tox-py37/6aa9830/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/27/654827/2/check/openstack-tox-docs/597b6f5/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/2/check/build-openstack-releasenotes/3180ddb/html/ : SUCCESS in 5m 23s\n- murano-congress-devstack http://logs.openstack.org/27/654827/2/check/murano-congress-devstack/4920703/ : SUCCESS in 39m 47s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/2/check/murano-rally-task/e95cc00/ : SUCCESS in 33m 28s\n- murano-tempest-api http://logs.openstack.org/27/654827/2/check/murano-tempest-api/90e1c45/ : SUCCESS in 39m 19s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/2/check/murano-tempest-cfapi/6b12a76/ : SUCCESS in 35m 42s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/2/check/murano-grenade/d6c388d/ : SUCCESS in 1h 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f15a1a814db2141935a34a0e736683d46042234f","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-29 06:58:06.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"f94657ef0bcb89852579a4cf1138a50a3b7315a6","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-30 05:46:22.000000000","message":"Patch Set 2: Code-Review-1\n\nthere are so many place still use git.openstack.org and review.openstack.org, please check","accounts_in_message":[],"_revision_number":2},{"id":"6f92bb129412b2803546b6946b1c7f1fc3839076","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-04 07:25:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"691d4a01168a65024340299337ea6e8a4a5cea93","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-04 07:44:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95bae6dc4473c7b9e3a0a6fec9e906d00e623d43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:24:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/4/check/build-openstack-api-ref/8a1edf3/ : POST_FAILURE in 4m 26s\n- test-release-openstack http://logs.openstack.org/27/654827/4/check/test-release-openstack/5eaa311/ : SUCCESS in 3m 29s\n- openstack-tox-cover http://logs.openstack.org/27/654827/4/check/openstack-tox-cover/dce2c81/cover/ : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/4/check/openstack-tox-lower-constraints/af010d3/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/4/check/openstack-tox-pep8/1cf5812/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/4/check/openstack-tox-py27/376b783/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/4/check/openstack-tox-py36/6da2503/ : RETRY_LIMIT in 1m 42s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/4/check/openstack-tox-py37/2c294ba/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/27/654827/4/check/openstack-tox-docs/f499b59/html/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/4/check/build-openstack-releasenotes/413b920/html/ : SUCCESS in 5m 51s\n- murano-congress-devstack http://logs.openstack.org/27/654827/4/check/murano-congress-devstack/f53b8bb/ : SUCCESS in 36m 14s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/4/check/murano-rally-task/c37cac1/ : SUCCESS in 35m 52s\n- murano-tempest-api http://logs.openstack.org/27/654827/4/check/murano-tempest-api/d2fefd3/ : SUCCESS in 44m 54s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/4/check/murano-tempest-cfapi/f29e277/ : SUCCESS in 35m 30s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/4/check/murano-grenade/7dddd28/ : SUCCESS in 1h 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d11426a225e9923a32878cd0638154fa50e6b1f6","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 04:18:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"07189de1fed8e77773d6173c0146040e4033294b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 05:55:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/4/check/build-openstack-api-ref/14054d8/ : POST_FAILURE in 4m 13s\n- test-release-openstack http://logs.openstack.org/27/654827/4/check/test-release-openstack/ba108dd/ : SUCCESS in 3m 15s\n- openstack-tox-cover http://logs.openstack.org/27/654827/4/check/openstack-tox-cover/bd42eff/cover/ : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/4/check/openstack-tox-lower-constraints/248a56a/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/4/check/openstack-tox-pep8/ac9e350/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/4/check/openstack-tox-py27/534f73f/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/4/check/openstack-tox-py36/318d4b7/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/4/check/openstack-tox-py37/dbefceb/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/27/654827/4/check/openstack-tox-docs/fec1e8e/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/4/check/build-openstack-releasenotes/6040cf0/html/ : SUCCESS in 5m 39s\n- murano-congress-devstack http://logs.openstack.org/27/654827/4/check/murano-congress-devstack/ecd42d3/ : SUCCESS in 36m 43s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/4/check/murano-rally-task/130e26e/ : SUCCESS in 35m 54s\n- murano-tempest-api http://logs.openstack.org/27/654827/4/check/murano-tempest-api/2c961ab/ : SUCCESS in 36m 52s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/4/check/murano-tempest-cfapi/3e564a4/ : SUCCESS in 47m 02s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/4/check/murano-grenade/5aeb29b/ : SUCCESS in 1h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24899d8f4c6ce20a77659bbcd36c67522ad465a3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 06:02:42.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4cd1dd882064cdb26eb518272aa1d6c930733c57","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-25 03:44:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"30cc566135074c40364bab40ef572f0406426aea","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-25 03:50:11.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"32977f70cfdb3adcb5247fafae42f604985c4c80","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-25 03:54:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7480902d7a983ba8ce469652c9018c19bce12d90","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-25 06:21:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nalmost LGTM, pls update the another one, thanks","accounts_in_message":[],"_revision_number":6},{"id":"aa280aac811bd38b837e4e3b47ee57534e732dad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 06:23:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/6/check/build-openstack-api-ref/9a39f5a/html/ : SUCCESS in 2m 27s\n- test-release-openstack http://logs.openstack.org/27/654827/6/check/test-release-openstack/de306d3/ : SUCCESS in 1m 46s\n- openstack-tox-cover http://logs.openstack.org/27/654827/6/check/openstack-tox-cover/40438ad/ : RETRY_LIMIT in 2m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/6/check/openstack-tox-lower-constraints/e03c9bd/ : RETRY_LIMIT in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/6/check/openstack-tox-pep8/e762a63/ : RETRY_LIMIT in 2m 28s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/6/check/openstack-tox-py27/2e754d5/ : RETRY_LIMIT in 2m 14s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/6/check/openstack-tox-py36/100c689/ : RETRY_LIMIT in 1m 31s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/6/check/openstack-tox-py37/5cad431/ : RETRY_LIMIT in 2m 18s\n- openstack-tox-docs http://logs.openstack.org/27/654827/6/check/openstack-tox-docs/70dbb3a/html/ : SUCCESS in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/6/check/build-openstack-releasenotes/3f21d05/html/ : SUCCESS in 4m 50s\n- murano-congress-devstack http://logs.openstack.org/27/654827/6/check/murano-congress-devstack/8f18c5d/ : SUCCESS in 38m 16s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/6/check/murano-rally-task/3e4413c/ : SUCCESS in 40m 23s\n- murano-tempest-api http://logs.openstack.org/27/654827/6/check/murano-tempest-api/edeec6a/ : SUCCESS in 45m 03s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/6/check/murano-tempest-cfapi/97dc653/ : SUCCESS in 32m 18s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/6/check/murano-grenade/cd47e01/ : SUCCESS in 1h 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f02c3d5897cc1972b63b324b21d971d278cbe924","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-25 06:43:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"db5d29281c2a702d41f63065d232232f41c43d8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 09:33:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/check/build-openstack-api-ref/2dacd33/html/ : SUCCESS in 2m 49s\n- test-release-openstack http://logs.openstack.org/27/654827/7/check/test-release-openstack/bcedef1/ : SUCCESS in 2m 12s\n- openstack-tox-cover http://logs.openstack.org/27/654827/7/check/openstack-tox-cover/f269cf9/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/check/openstack-tox-lower-constraints/aff3e91/ : RETRY_LIMIT in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/aacd2ca/ : RETRY_LIMIT in 1m 39s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/check/openstack-tox-py27/afa1eb2/ : RETRY_LIMIT in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/check/openstack-tox-py36/33b8102/ : RETRY_LIMIT in 1m 57s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/check/openstack-tox-py37/2d613d4/ : RETRY_LIMIT in 2m 10s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/check/openstack-tox-docs/dbdd871/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/check/build-openstack-releasenotes/caa1406/html/ : SUCCESS in 5m 56s\n- murano-congress-devstack http://logs.openstack.org/27/654827/7/check/murano-congress-devstack/d49afa1/ : SUCCESS in 38m 01s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/7/check/murano-rally-task/ab47a67/ : SUCCESS in 37m 58s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/check/murano-tempest-api/53d9451/ : SUCCESS in 43m 18s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/7/check/murano-tempest-cfapi/c156251/ : SUCCESS in 35m 46s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/7/check/murano-grenade/17add3f/ : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"56f908b11b005e75a3db3d0bd5b72406d9104567","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-26 01:47:45.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"a482426e0338cdc05354ad123f031aeb5ff86ffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 02:48:15.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/check/build-openstack-api-ref/02458ff/html/ : SUCCESS in 2m 44s\n- test-release-openstack http://logs.openstack.org/27/654827/7/check/test-release-openstack/f320ff9/ : SUCCESS in 1m 49s\n- openstack-tox-cover http://logs.openstack.org/27/654827/7/check/openstack-tox-cover/6903333/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/check/openstack-tox-lower-constraints/994c91e/ : RETRY_LIMIT in 2m 10s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/f2728d1/ : RETRY_LIMIT in 2m 08s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/check/openstack-tox-py27/27c7567/ : RETRY_LIMIT in 1m 59s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/check/openstack-tox-py36/9c3157c/ : RETRY_LIMIT in 2m 12s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/check/openstack-tox-py37/4cd0683/ : RETRY_LIMIT in 2m 55s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/check/openstack-tox-docs/e19c87d/html/ : SUCCESS in 3m 18s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/check/build-openstack-releasenotes/d9498a6/html/ : SUCCESS in 4m 06s\n- murano-congress-devstack http://logs.openstack.org/27/654827/7/check/murano-congress-devstack/f3aa4d4/ : SUCCESS in 39m 10s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/7/check/murano-rally-task/68a7870/ : SUCCESS in 42m 09s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/check/murano-tempest-api/8f288d5/ : SUCCESS in 47m 46s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/7/check/murano-tempest-cfapi/705a2c9/ : SUCCESS in 35m 35s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/7/check/murano-grenade/55c1004/ : FAILURE in 57m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f42e57fed3ce9f4882485b115bfb99900293b23a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-27 06:13:52.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"45512ccf828a1410bb82430fd06dbd798ff44850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:34:43.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/check/build-openstack-api-ref/e641ac9/html/ : SUCCESS in 3m 43s\n- test-release-openstack http://logs.openstack.org/27/654827/7/check/test-release-openstack/2036a50/ : SUCCESS in 2m 44s\n- openstack-tox-cover http://logs.openstack.org/27/654827/7/check/openstack-tox-cover/99f52da/ : RETRY_LIMIT in 2m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/check/openstack-tox-lower-constraints/dabd844/ : RETRY_LIMIT in 2m 09s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/d025900/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/check/openstack-tox-py27/2c929b7/ : RETRY_LIMIT in 1m 47s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/check/openstack-tox-py36/d423477/ : RETRY_LIMIT in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/check/openstack-tox-py37/2050372/ : RETRY_LIMIT in 1m 50s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/check/openstack-tox-docs/b5e7c40/html/ : SUCCESS in 2m 56s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/check/build-openstack-releasenotes/4cf9906/html/ : SUCCESS in 4m 43s\n- murano-congress-devstack http://logs.openstack.org/27/654827/7/check/murano-congress-devstack/c913f14/ : SUCCESS in 35m 17s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/7/check/murano-rally-task/4a85898/ : SUCCESS in 35m 56s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/check/murano-tempest-api/3f9a5a5/ : SUCCESS in 59m 36s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/7/check/murano-tempest-cfapi/628fa90/ : SUCCESS in 35m 14s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/7/check/murano-grenade/196dcf3/ : FAILURE in 50m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"47cd7ba37a68c4070a1975da77ae9e75d60b68c2","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-27 11:52:50.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"a87ee664c29a1b623dbdf19e528bf62930e85f96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:18:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/check/build-openstack-api-ref/abcc2dc/html/ : SUCCESS in 2m 57s\n- test-release-openstack http://logs.openstack.org/27/654827/7/check/test-release-openstack/ced9f08/ : SUCCESS in 2m 00s\n- openstack-tox-cover http://logs.openstack.org/27/654827/7/check/openstack-tox-cover/b7daba5/ : RETRY_LIMIT in 2m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/check/openstack-tox-lower-constraints/ccc86f1/ : RETRY_LIMIT in 1m 58s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/580cace/ : RETRY_LIMIT in 1m 58s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/check/openstack-tox-py27/aaa9856/ : RETRY_LIMIT in 2m 03s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/check/openstack-tox-py36/f0e4549/ : RETRY_LIMIT in 2m 08s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/check/openstack-tox-py37/38a2761/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/check/openstack-tox-docs/3dbcd80/html/ : SUCCESS in 4m 01s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/check/build-openstack-releasenotes/17f464f/html/ : SUCCESS in 4m 58s\n- murano-congress-devstack http://logs.openstack.org/27/654827/7/check/murano-congress-devstack/0e0a35d/ : SUCCESS in 40m 36s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/7/check/murano-rally-task/dd72f15/ : SUCCESS in 37m 42s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/check/murano-tempest-api/d1cd405/ : SUCCESS in 42m 47s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/7/check/murano-tempest-cfapi/62603b3/ : SUCCESS in 38m 22s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/7/check/murano-grenade/b39a661/ : FAILURE in 1h 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1f47502ad7e9646cd3f9c1d60e4b4ca0b53f6cef","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-28 02:13:48.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"04bbf7f6a4c14d88a96a3ef25918b569946b0671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 03:29:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/check/build-openstack-api-ref/1327835/html/ : SUCCESS in 3m 42s\n- test-release-openstack http://logs.openstack.org/27/654827/7/check/test-release-openstack/0c189fc/ : SUCCESS in 2m 26s\n- openstack-tox-cover http://logs.openstack.org/27/654827/7/check/openstack-tox-cover/f66df33/cover/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/check/openstack-tox-lower-constraints/4e89a6a/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/4b7187f/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/check/openstack-tox-py27/7cc60c4/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/check/openstack-tox-py36/61ba6f6/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/check/openstack-tox-py37/ff8c09f/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/check/openstack-tox-docs/137bbae/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/check/build-openstack-releasenotes/b7af6f0/html/ : SUCCESS in 5m 39s\n- murano-congress-devstack http://logs.openstack.org/27/654827/7/check/murano-congress-devstack/bbefd80/ : SUCCESS in 37m 45s (non-voting)\n- murano-rally-task http://logs.openstack.org/27/654827/7/check/murano-rally-task/3228178/ : SUCCESS in 39m 51s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/check/murano-tempest-api/e66a127/ : SUCCESS in 59m 40s\n- murano-tempest-cfapi http://logs.openstack.org/27/654827/7/check/murano-tempest-cfapi/e3e3d09/ : SUCCESS in 34m 25s (non-voting)\n- murano-grenade http://logs.openstack.org/27/654827/7/check/murano-grenade/c50b5ed/ : SUCCESS in 1h 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9db103818b0e0a78701f20ab57014a0f795008e8","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-28 03:30:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6a46f26484138d043e257c7a18bd61d206e38d91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 03:30:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"54bc8f92685ffe235cf4f3ee4c6763c87c241a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 04:25:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"be547dc9fff24b28fb905660672be3ce7e48a841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 04:25:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/27/654827/7/gate/build-openstack-api-ref/178085a/html/ : SUCCESS in 4m 06s\n- test-release-openstack http://logs.openstack.org/27/654827/7/gate/test-release-openstack/d0141f3/ : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/654827/7/gate/openstack-tox-lower-constraints/2e7b1ee/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/27/654827/7/gate/openstack-tox-pep8/3ee348c/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/27/654827/7/gate/openstack-tox-py27/d2fc4ef/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/27/654827/7/gate/openstack-tox-py36/276e227/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/27/654827/7/gate/openstack-tox-py37/48434e9/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/27/654827/7/gate/openstack-tox-docs/3cd8e2f/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/27/654827/7/gate/build-openstack-releasenotes/ae942ca/html/ : SUCCESS in 6m 15s\n- murano-tempest-api http://logs.openstack.org/27/654827/7/gate/murano-tempest-api/58e7556/ : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ecf8effdfae208e86133e329f5c6a514cf346bc4","revisions":{"cbb1d30284fd0dbb5ffc8250ab8be482815b07bb":{"kind":"REWORK","_number":1,"created":"2019-04-23 08:15:34.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/1"}}},"commit":{"parents":[{"commit":"d26b42250e6a005d7a3d12768f041a2bb1f33292","subject":"Remove the unused DISCOVER_DIRECTORY param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d26b42250e6a005d7a3d12768f041a2bb1f33292"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:15:25.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/cbb1d30284fd0dbb5ffc8250ab8be482815b07bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/cbb1d30284fd0dbb5ffc8250ab8be482815b07bb"}]},"branch":"refs/heads/master"},"a30ac8b3532018f2c7468cef5bc5474fa4d279e5":{"kind":"REWORK","_number":2,"created":"2019-04-24 08:06:08.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/2"}}},"commit":{"parents":[{"commit":"d26b42250e6a005d7a3d12768f041a2bb1f33292","subject":"Remove the unused DISCOVER_DIRECTORY param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d26b42250e6a005d7a3d12768f041a2bb1f33292"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-24 08:05:57.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/a30ac8b3532018f2c7468cef5bc5474fa4d279e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/a30ac8b3532018f2c7468cef5bc5474fa4d279e5"}]},"branch":"refs/heads/master"},"5c63aa0b79e070033177bd8e1c42b34d46b49ceb":{"kind":"REWORK","_number":3,"created":"2019-06-04 07:25:35.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/3"}}},"commit":{"parents":[{"commit":"8fb05fd9f4eba27c31b4a116dc771a53d96a4fb3","subject":"Merge \"Update the murano-rally-task job\u0027s irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8fb05fd9f4eba27c31b4a116dc771a53d96a4fb3"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-04 07:25:16.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5c63aa0b79e070033177bd8e1c42b34d46b49ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5c63aa0b79e070033177bd8e1c42b34d46b49ceb"}]},"branch":"refs/heads/master"},"bbf17e53464f60a3992f9cadda4b7bad10af2a7e":{"kind":"REWORK","_number":4,"created":"2019-06-04 07:44:38.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/4"}}},"commit":{"parents":[{"commit":"8fb05fd9f4eba27c31b4a116dc771a53d96a4fb3","subject":"Merge \"Update the murano-rally-task job\u0027s irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8fb05fd9f4eba27c31b4a116dc771a53d96a4fb3"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-04 07:44:29.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/bbf17e53464f60a3992f9cadda4b7bad10af2a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/bbf17e53464f60a3992f9cadda4b7bad10af2a7e"}]},"branch":"refs/heads/master"},"de915e4fa3c0e65e3a82563841572fd451b1cdaf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-25 03:44:45.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/27/654827/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/5"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-25 03:44:45.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/de915e4fa3c0e65e3a82563841572fd451b1cdaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/de915e4fa3c0e65e3a82563841572fd451b1cdaf"}]},"branch":"refs/heads/master"},"afda6e1c0ea8b1461ad59e4a04a5742ff547b496":{"kind":"REWORK","_number":6,"created":"2019-06-25 03:54:15.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/6"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-25 03:53:44.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/afda6e1c0ea8b1461ad59e4a04a5742ff547b496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/afda6e1c0ea8b1461ad59e4a04a5742ff547b496"}]},"branch":"refs/heads/master"},"ecf8effdfae208e86133e329f5c6a514cf346bc4":{"kind":"REWORK","_number":7,"created":"2019-06-25 06:43:50.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/27/654827/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/27/654827/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/27/654827/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/27/654827/7"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:14:33.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-25 06:43:40.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org counterparts.\n\nChange-Id: Id506aad04b878ba516872ef0169664f47f6eaab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ecf8effdfae208e86133e329f5c6a514cf346bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ecf8effdfae208e86133e329f5c6a514cf346bc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
