)]}'
{"id":"openstack%2Fmurano~658812","triplet_id":"openstack%2Fmurano~master~I8873d3ae033a4b4608e5bc050836102f55655142","project":"openstack/murano","branch":"master","topic":"bandit","hashtags":[],"change_id":"I8873d3ae033a4b4608e5bc050836102f55655142","subject":"Fix Sphinx lower-constraints","status":"ABANDONED","created":"2019-05-13 13:21:59.000000000","updated":"2019-05-14 09:00:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f043307ee1b8c824075c7222af9a9a823f2ebea","_number":658812,"virtual_id_number":658812,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-13 15:03:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 15:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d29b2373b465dadcdadd17ec755cab483b95796f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 13:21:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32bd20c52d8286b8ea48c6666f306a548282237a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 15:03:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/658812/1/check/requirements-check/04a5cd8/ : FAILURE in 10m 21s\n- openstack-tox-cover http://logs.openstack.org/12/658812/1/check/openstack-tox-cover/7175083/cover/ : SUCCESS in 14m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/658812/1/check/openstack-tox-lower-constraints/9a1edd5/ : SUCCESS in 14m 37s\n- openstack-tox-pep8 http://logs.openstack.org/12/658812/1/check/openstack-tox-pep8/215fedc/ : FAILURE in 15m 37s\n- openstack-tox-py27 http://logs.openstack.org/12/658812/1/check/openstack-tox-py27/28da7ff/ : SUCCESS in 16m 13s\n- openstack-tox-py36 http://logs.openstack.org/12/658812/1/check/openstack-tox-py36/0a87d08/ : SUCCESS in 14m 28s\n- openstack-tox-py37 http://logs.openstack.org/12/658812/1/check/openstack-tox-py37/e44aa82/ : SUCCESS in 18m 48s\n- openstack-tox-docs http://logs.openstack.org/12/658812/1/check/openstack-tox-docs/0faacfb/html/ : SUCCESS in 4m 33s\n- murano-congress-devstack http://logs.openstack.org/12/658812/1/check/murano-congress-devstack/0f65f3e/ : SUCCESS in 50m 12s (non-voting)\n- murano-rally-task http://logs.openstack.org/12/658812/1/check/murano-rally-task/0f502ca/ : SUCCESS in 49m 15s\n- murano-tempest-api http://logs.openstack.org/12/658812/1/check/murano-tempest-api/aef9f0c/ : SUCCESS in 59m 25s\n- murano-tempest-cfapi http://logs.openstack.org/12/658812/1/check/murano-tempest-cfapi/13605ad/ : SUCCESS in 49m 55s (non-voting)\n- murano-grenade http://logs.openstack.org/12/658812/1/check/murano-grenade/400009f/ : SUCCESS in 1h 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"053e9f27dcd8b534982aaf6d863a607e3a12a5ec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 09:00:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ded21e5f298d2963bc4337e5402bba06ed6f3bf","revisions":{"8ded21e5f298d2963bc4337e5402bba06ed6f3bf":{"kind":"REWORK","_number":1,"created":"2019-05-13 13:21:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/12/658812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/12/658812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/12/658812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/12/658812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/12/658812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/12/658812/1"}}},"commit":{"parents":[{"commit":"bbbb73989297d377a78c015920ebcc6085961744","subject":"Merge \"Add framework for murano-status upgrade check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/bbbb73989297d377a78c015920ebcc6085961744"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-13 13:17:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-13 13:17:04.000000000","tz":120},"subject":"Fix Sphinx lower-constraints","message":"Fix Sphinx lower-constraints\n\nRelated to Ibaa10b092b262e49435d35f093906bb75dc9762f\n\nChange-Id: I8873d3ae033a4b4608e5bc050836102f55655142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ded21e5f298d2963bc4337e5402bba06ed6f3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ded21e5f298d2963bc4337e5402bba06ed6f3bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
