)]}'
{"id":"openstack%2Fmurano~667410","triplet_id":"openstack%2Fmurano~master~I37d64aca485eb81ac16d7d85d5b695e573c8793c","project":"openstack/murano","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"I37d64aca485eb81ac16d7d85d5b695e573c8793c","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-25 16:56:22.000000000","updated":"2019-06-28 01:41:20.000000000","submitted":"2019-06-28 01:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667410-1561686080425-be057232","meta_rev_id":"5ebcafbb06f6ac3ad1a4548251e1f81662a802b5","_number":667410,"virtual_id_number":667410,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-27 13:12:51.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-28 01:41:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-27 13:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-06-28 00:52:39.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-06-28 00:52:39.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 13:00:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-28 00:52:39.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-06-28 01:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56ba5adb721929a0ef076c61c69060132b8779f0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-25 16:56:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00218a0b3b56c9b6f0ed862fa338ff44bc749e04","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-25 16:59:59.000000000","message":"Patch Set 1:\n\nWithout this patch some tests fail [1] with \u0027command not found\u0027 errors.\n\n[1] http://logs.openstack.org/27/654827/7/check/openstack-tox-pep8/aacd2ca/job-output.txt.gz#_2019-06-25_07_37_57_432199","accounts_in_message":[],"_revision_number":1},{"id":"86d14a3cbfe81f21ba44b12792aa78a2a48b01de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:32:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/1/check/build-openstack-api-ref/38da487/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/10/667410/1/check/openstack-tox-cover/98889bc/ : RETRY_LIMIT in 2m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/1/check/openstack-tox-lower-constraints/f5d1a55/ : RETRY_LIMIT in 2m 11s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/1/check/openstack-tox-pep8/14aab5d/ : RETRY_LIMIT in 1m 58s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/1/check/openstack-tox-py27/30d4506/ : RETRY_LIMIT in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/1/check/openstack-tox-py36/e5f283d/ : RETRY_LIMIT in 2m 14s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/1/check/openstack-tox-py37/0749c8f/ : RETRY_LIMIT in 2m 01s\n- openstack-tox-docs http://logs.openstack.org/10/667410/1/check/openstack-tox-docs/5e52b6b/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/1/check/build-openstack-releasenotes/b3886dc/html/ : SUCCESS in 4m 54s\n- murano-congress-devstack http://logs.openstack.org/10/667410/1/check/murano-congress-devstack/712876e/ : SUCCESS in 38m 05s (non-voting)\n- murano-rally-task http://logs.openstack.org/10/667410/1/check/murano-rally-task/33d4e18/ : SUCCESS in 36m 57s\n- murano-tempest-api http://logs.openstack.org/10/667410/1/check/murano-tempest-api/36805ad/ : SUCCESS in 43m 28s\n- murano-tempest-cfapi http://logs.openstack.org/10/667410/1/check/murano-tempest-cfapi/81ef75e/ : SUCCESS in 38m 26s (non-voting)\n- murano-grenade http://logs.openstack.org/10/667410/1/check/murano-grenade/831c0ca/ : SUCCESS in 1h 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79d52bb12a6344953289b5c2d1034e81e3a4319a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 06:20:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"131af98151f20cf0b094b20ff9013973a68dcd74","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 07:26:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6711830f796fb7e0bce0bd425139b22b1e93b6da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:19:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/3/check/build-openstack-api-ref/e5b8ac2/html/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/10/667410/3/check/openstack-tox-cover/a159407/cover/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/3/check/openstack-tox-lower-constraints/169149f/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/3/check/openstack-tox-pep8/ba36902/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/3/check/openstack-tox-py27/7fe2ca0/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/3/check/openstack-tox-py36/f2e3737/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/3/check/openstack-tox-py37/5766fb6/ : FAILURE in 2m 56s\n- openstack-tox-docs http://logs.openstack.org/10/667410/3/check/openstack-tox-docs/af44766/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/3/check/build-openstack-releasenotes/15be6d7/html/ : SUCCESS in 5m 03s\n- murano-congress-devstack http://logs.openstack.org/10/667410/3/check/murano-congress-devstack/62be952/ : SUCCESS in 46m 26s (non-voting)\n- murano-rally-task http://logs.openstack.org/10/667410/3/check/murano-rally-task/3eb6341/ : SUCCESS in 33m 48s\n- murano-tempest-api http://logs.openstack.org/10/667410/3/check/murano-tempest-api/cea782e/ : SUCCESS in 44m 55s\n- murano-tempest-cfapi http://logs.openstack.org/10/667410/3/check/murano-tempest-cfapi/04813e7/ : SUCCESS in 31m 55s (non-voting)\n- murano-grenade http://logs.openstack.org/10/667410/3/check/murano-grenade/e4e0eaf/ : FAILURE in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"936fa5853cfabb130512bf391abceaaffb4e6c9a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 08:40:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ea27154685ac580514badacc72c46841e233b620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:00:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/3/check/build-openstack-api-ref/a610707/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover http://logs.openstack.org/10/667410/3/check/openstack-tox-cover/7f4b87b/cover/ : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/3/check/openstack-tox-lower-constraints/bd64736/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/3/check/openstack-tox-pep8/956be6d/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/3/check/openstack-tox-py27/0136a12/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/3/check/openstack-tox-py36/8edb0bb/ : SUCCESS in 5m 59s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/3/check/openstack-tox-py37/ed32dbb/ : FAILURE in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/10/667410/3/check/openstack-tox-docs/5117573/html/ : SUCCESS in 5m 11s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/3/check/build-openstack-releasenotes/31961d3/html/ : SUCCESS in 6m 02s\n- murano-congress-devstack http://logs.openstack.org/10/667410/3/check/murano-congress-devstack/6c8babb/ : SUCCESS in 39m 41s (non-voting)\n- murano-rally-task http://logs.openstack.org/10/667410/3/check/murano-rally-task/d4e3fb9/ : SUCCESS in 37m 52s\n- murano-tempest-api http://logs.openstack.org/10/667410/3/check/murano-tempest-api/b88afca/ : SUCCESS in 45m 25s\n- murano-tempest-cfapi http://logs.openstack.org/10/667410/3/check/murano-tempest-cfapi/1bf50fb/ : SUCCESS in 39m 06s (non-voting)\n- murano-grenade http://logs.openstack.org/10/667410/3/check/murano-grenade/a9ce3a6/ : SUCCESS in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9803450152816d183d827a8547b864a080ed5da","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 10:40:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"436ad896f28ee28d3f72b05a88a3d05286815159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:14:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/4/check/build-openstack-api-ref/4f9879f/html/ : SUCCESS in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/10/667410/4/check/openstack-tox-cover/217c35f/cover/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/4/check/openstack-tox-lower-constraints/cb80bf7/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/4/check/openstack-tox-pep8/dedd4e9/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/4/check/openstack-tox-py27/f1efcae/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/4/check/openstack-tox-py36/c2184bb/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/4/check/openstack-tox-py37/c7c6ede/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/10/667410/4/check/openstack-tox-docs/5b7b34a/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/4/check/build-openstack-releasenotes/21a515d/ : RETRY_LIMIT in 2m 15s\n- murano-congress-devstack http://logs.openstack.org/10/667410/4/check/murano-congress-devstack/32add4d/ : TIMED_OUT in 1h 13m 04s (non-voting)\n- murano-rally-task http://logs.openstack.org/10/667410/4/check/murano-rally-task/82e4569/ : SUCCESS in 33m 58s\n- murano-tempest-api http://logs.openstack.org/10/667410/4/check/murano-tempest-api/325ad43/ : SUCCESS in 45m 06s\n- murano-tempest-cfapi http://logs.openstack.org/10/667410/4/check/murano-tempest-cfapi/4b5ff75/ : SUCCESS in 33m 16s (non-voting)\n- murano-grenade http://logs.openstack.org/10/667410/4/check/murano-grenade/2755ef6/ : SUCCESS in 1h 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8487d2f8e21c9eaf54279a38f8d715ffa05ce29f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-27 03:14:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0872c3ccdbaca546125f2fe9a1ca91ad746d18d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:49:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/4/check/build-openstack-api-ref/be53258/html/ : SUCCESS in 3m 14s\n- openstack-tox-cover http://logs.openstack.org/10/667410/4/check/openstack-tox-cover/c6c1b98/cover/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/4/check/openstack-tox-lower-constraints/084c735/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/4/check/openstack-tox-pep8/24de8a2/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/4/check/openstack-tox-py27/9a1ccf7/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/4/check/openstack-tox-py36/ac09553/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/4/check/openstack-tox-py37/5989242/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/10/667410/4/check/openstack-tox-docs/fc8c988/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/4/check/build-openstack-releasenotes/ca88da5/html/ : SUCCESS in 5m 21s\n- murano-congress-devstack http://logs.openstack.org/10/667410/4/check/murano-congress-devstack/0432fc3/ : SUCCESS in 40m 58s (non-voting)\n- murano-rally-task http://logs.openstack.org/10/667410/4/check/murano-rally-task/2706b54/ : SUCCESS in 37m 08s\n- murano-tempest-api http://logs.openstack.org/10/667410/4/check/murano-tempest-api/5b4b106/ : SUCCESS in 41m 53s\n- murano-tempest-cfapi http://logs.openstack.org/10/667410/4/check/murano-tempest-cfapi/6498b13/ : SUCCESS in 33m 58s (non-voting)\n- murano-grenade http://logs.openstack.org/10/667410/4/check/murano-grenade/c4cc0ed/ : SUCCESS in 1h 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08a7655c4ad26af3cd3d73af2eab7821047afeb8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 13:00:44.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30fa323be1145b50977860e8bbc7d4296c47d041","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 13:12:51.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThanks Sean for the reviews!","accounts_in_message":[],"_revision_number":4},{"id":"631317d209e52cb55cebddc825bcf4c932aadca5","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-28 00:52:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":4},{"id":"f8273a53a899cdae0d3b6d0e3b9c32ee425bb257","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 00:52:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a924924306a742751bfd798dec7de165997c8621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:41:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d7094211f964f28c3cea07bc0d4ba8396a4d1ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:41:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/10/667410/4/gate/build-openstack-api-ref/012aea1/html/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/667410/4/gate/openstack-tox-lower-constraints/effc615/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/10/667410/4/gate/openstack-tox-pep8/85e0f28/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/10/667410/4/gate/openstack-tox-py27/d69448e/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/10/667410/4/gate/openstack-tox-py36/cac8054/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/10/667410/4/gate/openstack-tox-py37/db29d2c/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/10/667410/4/gate/openstack-tox-docs/008d88a/html/ : SUCCESS in 4m 04s\n- build-openstack-releasenotes http://logs.openstack.org/10/667410/4/gate/build-openstack-releasenotes/8267fb5/html/ : SUCCESS in 5m 48s\n- murano-tempest-api http://logs.openstack.org/10/667410/4/gate/murano-tempest-api/33cb3ad/ : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ddd5378ac8535b48e5ae2885b3187a5b1b88b766","revisions":{"cd6401887f20f9978c8ae63dd9e6b1a8040202d8":{"kind":"REWORK","_number":1,"created":"2019-06-25 16:56:22.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/667410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/10/667410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/10/667410/1"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-25 16:53:50.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-25 16:56:20.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I37d64aca485eb81ac16d7d85d5b695e573c8793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/cd6401887f20f9978c8ae63dd9e6b1a8040202d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/cd6401887f20f9978c8ae63dd9e6b1a8040202d8"}]},"branch":"refs/heads/master"},"f4cef951475c0b297f4e8319124436cf7ef6933d":{"kind":"REWORK","_number":2,"created":"2019-06-26 06:20:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/667410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/10/667410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/10/667410/2"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-25 16:53:50.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 06:19:58.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I37d64aca485eb81ac16d7d85d5b695e573c8793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f4cef951475c0b297f4e8319124436cf7ef6933d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f4cef951475c0b297f4e8319124436cf7ef6933d"}]},"branch":"refs/heads/master"},"eba472c1b07bc2534b58ab392cf6c90b562d0953":{"kind":"REWORK","_number":3,"created":"2019-06-26 07:26:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/667410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/10/667410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/10/667410/3"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-25 16:53:50.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 07:25:44.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I37d64aca485eb81ac16d7d85d5b695e573c8793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/eba472c1b07bc2534b58ab392cf6c90b562d0953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/eba472c1b07bc2534b58ab392cf6c90b562d0953"}]},"branch":"refs/heads/master"},"ddd5378ac8535b48e5ae2885b3187a5b1b88b766":{"kind":"REWORK","_number":4,"created":"2019-06-26 10:40:56.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/667410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/10/667410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/10/667410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/10/667410/4"}}},"commit":{"parents":[{"commit":"808fc6408cbe61565d9504fac5a9c00b2abeefe6","subject":"Add deps for api-ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/808fc6408cbe61565d9504fac5a9c00b2abeefe6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-25 16:53:50.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 10:35:58.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn test jobs the script tools/test-setup.sh is called which requires\nmysql and postgres servers and clients to be installed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I37d64aca485eb81ac16d7d85d5b695e573c8793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ddd5378ac8535b48e5ae2885b3187a5b1b88b766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ddd5378ac8535b48e5ae2885b3187a5b1b88b766"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
