)]}'
{"id":"openstack%2Fmurano~669092","triplet_id":"openstack%2Fmurano~master~Id8e5adb6bd9b66318729cb5f222ee8dfb53e2e78","project":"openstack/murano","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Id8e5adb6bd9b66318729cb5f222ee8dfb53e2e78","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-07-04 08:05:44.000000000","updated":"2019-07-04 11:53:47.000000000","submitted":"2019-07-04 11:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669092-1562241227850-5870c8eb","meta_rev_id":"8b1622afbe709c00c7ce0a959d7100fe9982d26c","_number":669092,"virtual_id_number":669092,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-07-04 11:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-07-04 11:41:49.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-07-04 11:41:49.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 11:41:49.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-07-04 11:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c6751a98f73a836cc14e2337983209286cfabb6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 08:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bb953cd9e4a6f91635a6b508c2dd1f78959393d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:20:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/92/669092/1/check/build-openstack-api-ref/c5e11ec/html/ : SUCCESS in 3m 48s\n- requirements-check http://logs.openstack.org/92/669092/1/check/requirements-check/44988fc/ : SUCCESS in 2m 35s\n- openstack-tox-pep8 http://logs.openstack.org/92/669092/1/check/openstack-tox-pep8/22ffe58/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/92/669092/1/check/openstack-tox-docs/87295b3/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/92/669092/1/check/build-openstack-releasenotes/fdc4aad/html/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"beca1c4608d62c1795a4d51b774e3b8b4b239493","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 11:41:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"895f71aa52d83ff113d1ab944e20a000cd41d7b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:42:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d50c799f3bdd5db32cff5f014d05d69a856f52f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:53:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce8a31e383c81bd0a428776c128fc15436d76acc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/92/669092/1/gate/build-openstack-api-ref/efa699f/html/ : SUCCESS in 3m 40s\n- requirements-check http://logs.openstack.org/92/669092/1/gate/requirements-check/ba7c279/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/92/669092/1/gate/openstack-tox-pep8/0352ae9/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/92/669092/1/gate/openstack-tox-docs/c2f1153/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/92/669092/1/gate/build-openstack-releasenotes/b441424/html/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d816575020366debbe15f7d8f7729c916626f10","revisions":{"3d816575020366debbe15f7d8f7729c916626f10":{"kind":"REWORK","_number":1,"created":"2019-07-04 08:05:44.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/92/669092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/92/669092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/669092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/669092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/669092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/92/669092/1"}}},"commit":{"parents":[{"commit":"e9cde491c08caa49af0dadc38071ab0589e85465","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e9cde491c08caa49af0dadc38071ab0589e85465"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:04:35.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:05:46.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\nDepend-On: https://review.opendev.org/#/c/663060/\n\nChange-Id: Id8e5adb6bd9b66318729cb5f222ee8dfb53e2e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/3d816575020366debbe15f7d8f7729c916626f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/3d816575020366debbe15f7d8f7729c916626f10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
