)]}'
{"id":"openstack%2Fmurano~673597","triplet_id":"openstack%2Fmurano~master~I222964e47eebed6b3593b1954766a0691802ffef","project":"openstack/murano","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"I222964e47eebed6b3593b1954766a0691802ffef","subject":"Update api-ref location","status":"MERGED","created":"2019-07-30 17:52:09.000000000","updated":"2019-07-31 08:28:54.000000000","submitted":"2019-07-31 08:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673597-1564561734636-efd57ca1","meta_rev_id":"0feebb945a7036da4757629c498a732c28e425eb","_number":673597,"virtual_id_number":673597,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-07-31 08:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-07-31 07:32:52.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-07-31 07:32:52.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 07:32:52.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-07-31 08:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd4b41df594d7e5fb713914e7cb815ac0ce1c487","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-30 17:52:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d40fb1e6f2734f07f52395c8d2c25365099982a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:05:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/97/673597/1/check/build-openstack-api-ref/64b7624/html/ : SUCCESS in 5m 07s\n- openstack-tox-cover http://logs.openstack.org/97/673597/1/check/openstack-tox-cover/748f563/cover/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/673597/1/check/openstack-tox-lower-constraints/3a19d1d/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/97/673597/1/check/openstack-tox-pep8/04ae035/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/97/673597/1/check/openstack-tox-py27/4b8e72d/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/97/673597/1/check/openstack-tox-py36/61cd29f/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/97/673597/1/check/openstack-tox-py37/2a42362/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/97/673597/1/check/openstack-tox-docs/546d6c3/html/ : SUCCESS in 6m 19s\n- build-openstack-releasenotes http://logs.openstack.org/97/673597/1/check/build-openstack-releasenotes/e9d4e56/html/ : SUCCESS in 6m 45s\n- murano-congress-devstack http://logs.openstack.org/97/673597/1/check/murano-congress-devstack/29c67ea/ : SUCCESS in 42m 59s (non-voting)\n- murano-rally-task http://logs.openstack.org/97/673597/1/check/murano-rally-task/c6a64e7/ : SUCCESS in 38m 58s\n- murano-tempest-api http://logs.openstack.org/97/673597/1/check/murano-tempest-api/5a8c4b3/ : SUCCESS in 48m 04s\n- murano-tempest-cfapi http://logs.openstack.org/97/673597/1/check/murano-tempest-cfapi/776cefb/ : SUCCESS in 41m 26s (non-voting)\n- murano-grenade http://logs.openstack.org/97/673597/1/check/murano-grenade/f9c880c/ : SUCCESS in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aec7f2e09548cc1468f90a7101174a823c140ea7","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-31 07:32:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa768df6f1ced6520c81f7db26e13f53f52a32ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 07:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e8d8873aa08cf03316e2f55e6d48ddd041b56c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:28:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b84e83f2830ac0cca6a2e8a89b5c3a8d39f87b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:28:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/97/673597/1/gate/build-openstack-api-ref/69c2f2f/html/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/673597/1/gate/openstack-tox-lower-constraints/7fea253/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/97/673597/1/gate/openstack-tox-pep8/9d8fe23/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/97/673597/1/gate/openstack-tox-py27/23cd52c/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/97/673597/1/gate/openstack-tox-py36/2baf39a/ : SUCCESS in 6m 39s\n- openstack-tox-py37 http://logs.openstack.org/97/673597/1/gate/openstack-tox-py37/cd787b5/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/97/673597/1/gate/openstack-tox-docs/a5da50a/html/ : SUCCESS in 5m 20s\n- build-openstack-releasenotes http://logs.openstack.org/97/673597/1/gate/build-openstack-releasenotes/4c7dc28/html/ : SUCCESS in 6m 03s\n- murano-tempest-api http://logs.openstack.org/97/673597/1/gate/murano-tempest-api/5ee80fe/ : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f83cbd4a388af486513de30e2dba9191bbbba7e9","revisions":{"f83cbd4a388af486513de30e2dba9191bbbba7e9":{"kind":"REWORK","_number":1,"created":"2019-07-30 17:52:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/673597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/97/673597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/97/673597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/97/673597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/97/673597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/97/673597/1"}}},"commit":{"parents":[{"commit":"7435a2e7bd7bb48b03f957a2edde32ad42e03fb7","subject":"Merge \"Bump to hacking 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7435a2e7bd7bb48b03f957a2edde32ad42e03fb7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:52:00.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:52:00.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: I222964e47eebed6b3593b1954766a0691802ffef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f83cbd4a388af486513de30e2dba9191bbbba7e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f83cbd4a388af486513de30e2dba9191bbbba7e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
