)]}'
{"id":"openstack%2Fmurano~673751","triplet_id":"openstack%2Fmurano~master~Id21850ce27f596d7bacc54fdccbaca766b40b568","project":"openstack/murano","branch":"master","hashtags":[],"change_id":"Id21850ce27f596d7bacc54fdccbaca766b40b568","subject":"Fix IPV6 rabbit host","status":"MERGED","created":"2019-07-31 08:02:23.000000000","updated":"2019-08-08 10:26:53.000000000","submitted":"2019-08-08 10:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673751-1565260013279-a273883a","meta_rev_id":"c1c09b437e103922490047472b9ad704739b8724","_number":673751,"virtual_id_number":673751,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-08-08 10:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-08-08 09:23:25.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-08-08 09:23:25.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 10:26:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"811e1c1ea7cb8602b69ea5537f0ea6a47013cc8f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-31 08:02:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c901674d89b7ff3530de13a696fa613c14d8dbda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 09:22:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/51/673751/1/check/openstack-tox-cover/72056bb/cover/ : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/673751/1/check/openstack-tox-lower-constraints/d9b6adf/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/51/673751/1/check/openstack-tox-pep8/0e6f694/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/51/673751/1/check/openstack-tox-py27/036ae6f/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/51/673751/1/check/openstack-tox-py36/e969d57/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/51/673751/1/check/openstack-tox-py37/66a16d0/ : SUCCESS in 6m 39s\n- openstack-tox-docs http://logs.openstack.org/51/673751/1/check/openstack-tox-docs/d7e13f1/html/ : SUCCESS in 4m 41s\n- murano-congress-devstack http://logs.openstack.org/51/673751/1/check/murano-congress-devstack/13239c7/ : SUCCESS in 43m 20s (non-voting)\n- murano-rally-task http://logs.openstack.org/51/673751/1/check/murano-rally-task/9dd4ded/ : SUCCESS in 42m 51s\n- murano-tempest-api http://logs.openstack.org/51/673751/1/check/murano-tempest-api/abb6447/ : SUCCESS in 46m 39s\n- murano-tempest-cfapi http://logs.openstack.org/51/673751/1/check/murano-tempest-cfapi/abfd9da/ : SUCCESS in 37m 31s (non-voting)\n- murano-grenade http://logs.openstack.org/51/673751/1/check/murano-grenade/d852857/ : SUCCESS in 1h 12m 29s (non-voting)\n- murano-tempest-api-ipv6-only http://logs.openstack.org/51/673751/1/check/murano-tempest-api-ipv6-only/f947d0e/ : FAILURE in 49m 56s","accounts_in_message":[],"_revision_number":1},{"id":"65f55d481f5ffb84ad5709de2f484c8aa666d131","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-31 09:31:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e11598b34c20a79972da9e149cb1b1f248294f0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 09:39:11.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"703d0efc97aed76c26187a8870514a2720f2e40c","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-31 09:40:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"88b24f0bc3292334d01d7732b1d5a676f205709e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 11:04:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/51/673751/2/check/openstack-tox-cover/80b9eaa/cover/ : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/673751/2/check/openstack-tox-lower-constraints/2801ff6/ : SUCCESS in 6m 03s\n- openstack-tox-pep8 http://logs.openstack.org/51/673751/2/check/openstack-tox-pep8/ebab412/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/51/673751/2/check/openstack-tox-py27/730d367/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/51/673751/2/check/openstack-tox-py36/4a47c99/ : SUCCESS in 4m 41s\n- openstack-tox-py37 http://logs.openstack.org/51/673751/2/check/openstack-tox-py37/a4fe3ba/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/51/673751/2/check/openstack-tox-docs/57393f4/html/ : SUCCESS in 4m 00s\n- murano-congress-devstack http://logs.openstack.org/51/673751/2/check/murano-congress-devstack/416759e/ : SUCCESS in 38m 07s (non-voting)\n- murano-rally-task http://logs.openstack.org/51/673751/2/check/murano-rally-task/3e245f2/ : SUCCESS in 40m 24s\n- murano-tempest-api http://logs.openstack.org/51/673751/2/check/murano-tempest-api/b54e64c/ : SUCCESS in 48m 18s\n- murano-tempest-cfapi http://logs.openstack.org/51/673751/2/check/murano-tempest-cfapi/07c34c5/ : SUCCESS in 35m 35s (non-voting)\n- murano-grenade http://logs.openstack.org/51/673751/2/check/murano-grenade/4c56ff3/ : SUCCESS in 1h 16m 48s (non-voting)\n- murano-tempest-api-ipv6-only http://logs.openstack.org/51/673751/2/check/murano-tempest-api-ipv6-only/3552c46/ : FAILURE in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b27e1ed26fef9748d978ba54b7168de58ae285a2","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-01 00:55:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"66147217c8ebc833402276e5338bd84de16fa6bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 02:22:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/51/673751/2/check/openstack-tox-cover/3135394/cover/ : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/51/673751/2/check/openstack-tox-lower-constraints/ca9d47f/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://logs.opendev.org/51/673751/2/check/openstack-tox-pep8/2b69ab1/ : SUCCESS in 4m 15s\n- openstack-tox-py27 https://logs.opendev.org/51/673751/2/check/openstack-tox-py27/866d0cc/ : SUCCESS in 4m 32s\n- openstack-tox-py36 https://logs.opendev.org/51/673751/2/check/openstack-tox-py36/f6577f6/ : SUCCESS in 4m 44s\n- openstack-tox-py37 https://logs.opendev.org/51/673751/2/check/openstack-tox-py37/5dd487b/ : SUCCESS in 5m 18s\n- openstack-tox-docs https://logs.opendev.org/51/673751/2/check/openstack-tox-docs/1a72f78/html/ : SUCCESS in 4m 08s\n- murano-congress-devstack https://logs.opendev.org/51/673751/2/check/murano-congress-devstack/e380b87/ : SUCCESS in 37m 13s (non-voting)\n- murano-rally-task https://logs.opendev.org/51/673751/2/check/murano-rally-task/2c46974/ : SUCCESS in 33m 44s\n- murano-tempest-api https://logs.opendev.org/51/673751/2/check/murano-tempest-api/006d0d0/ : SUCCESS in 40m 16s\n- murano-tempest-cfapi https://logs.opendev.org/51/673751/2/check/murano-tempest-cfapi/65b9942/ : SUCCESS in 34m 02s (non-voting)\n- murano-grenade https://logs.opendev.org/51/673751/2/check/murano-grenade/ae09d73/ : FAILURE in 1h 25m 25s (non-voting)\n- murano-tempest-api-ipv6-only https://logs.opendev.org/51/673751/2/check/murano-tempest-api-ipv6-only/def1d58/ : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f0841a6b22107cb65a6aa7c382527e05e5332986","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-08 01:24:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3d716b6b1acdf27ada84f025406c3a9519c6a9e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:36:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/51/673751/3/check/openstack-tox-cover/0ca48cb/cover/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://logs.opendev.org/51/673751/3/check/openstack-tox-lower-constraints/34a682c/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://logs.opendev.org/51/673751/3/check/openstack-tox-pep8/d5d1d9e/ : SUCCESS in 4m 14s\n- openstack-tox-py27 https://logs.opendev.org/51/673751/3/check/openstack-tox-py27/2239251/ : SUCCESS in 5m 09s\n- openstack-tox-py36 https://logs.opendev.org/51/673751/3/check/openstack-tox-py36/58f117c/ : SUCCESS in 4m 53s\n- openstack-tox-py37 https://logs.opendev.org/51/673751/3/check/openstack-tox-py37/8c3443b/ : SUCCESS in 4m 39s\n- openstack-tox-docs https://logs.opendev.org/51/673751/3/check/openstack-tox-docs/0e1e6aa/html/ : SUCCESS in 4m 46s\n- murano-congress-devstack https://logs.opendev.org/51/673751/3/check/murano-congress-devstack/0a2c854/ : SUCCESS in 41m 26s (non-voting)\n- murano-rally-task https://logs.opendev.org/51/673751/3/check/murano-rally-task/d05eb79/ : SUCCESS in 37m 04s\n- murano-tempest-api https://logs.opendev.org/51/673751/3/check/murano-tempest-api/0cf7133/ : SUCCESS in 41m 53s\n- murano-tempest-cfapi https://logs.opendev.org/51/673751/3/check/murano-tempest-cfapi/bbde31c/ : SUCCESS in 44m 56s (non-voting)\n- murano-grenade https://logs.opendev.org/51/673751/3/check/murano-grenade/9a15a18/ : SUCCESS in 1h 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cce2206edd660af1ecad1be5f87dcba6c8c8cd9a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-08 09:23:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a0d4656b3ebdfd431fd7e8b00baec665f4a71492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:23:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d54f9b09af4a99a8f02d49c1931e1ecbd79458a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:26:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1c09b437e103922490047472b9ad704739b8724","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:26:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/51/673751/3/gate/openstack-tox-lower-constraints/91b47d7/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://logs.opendev.org/51/673751/3/gate/openstack-tox-pep8/119a3b0/ : SUCCESS in 4m 45s\n- openstack-tox-py27 https://logs.opendev.org/51/673751/3/gate/openstack-tox-py27/3922c01/ : SUCCESS in 9m 23s\n- openstack-tox-py36 https://logs.opendev.org/51/673751/3/gate/openstack-tox-py36/a987820/ : SUCCESS in 7m 49s\n- openstack-tox-py37 https://logs.opendev.org/51/673751/3/gate/openstack-tox-py37/564359b/ : SUCCESS in 12m 25s\n- openstack-tox-docs https://logs.opendev.org/51/673751/3/gate/openstack-tox-docs/abb0452/html/ : SUCCESS in 7m 42s\n- murano-tempest-api https://logs.opendev.org/51/673751/3/gate/murano-tempest-api/3fa01c4/ : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c7fa9048c51c81e7135429bd0f3d49b55b07263","revisions":{"7af1be9bdcf502a96ee3ff2d76783616ca7e7053":{"kind":"REWORK","_number":1,"created":"2019-07-31 08:02:23.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/51/673751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/51/673751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/51/673751/1"}}},"commit":{"parents":[{"commit":"7435a2e7bd7bb48b03f957a2edde32ad42e03fb7","subject":"Merge \"Bump to hacking 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7435a2e7bd7bb48b03f957a2edde32ad42e03fb7"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-31 07:55:23.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-31 07:55:23.000000000","tz":480},"subject":"Fix IPV6","message":"Fix IPV6\n\nDepends-On: https://review.opendev.org/#/c/673291\nChange-Id: Id21850ce27f596d7bacc54fdccbaca766b40b568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7af1be9bdcf502a96ee3ff2d76783616ca7e7053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7af1be9bdcf502a96ee3ff2d76783616ca7e7053"}]},"branch":"refs/heads/master"},"5ec34722ee805024941d4e34634d1ebdf5275407":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-31 09:31:02.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/51/673751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/51/673751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/51/673751/2"}}},"commit":{"parents":[{"commit":"7435a2e7bd7bb48b03f957a2edde32ad42e03fb7","subject":"Merge \"Bump to hacking 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7435a2e7bd7bb48b03f957a2edde32ad42e03fb7"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-31 07:55:23.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-31 09:30:57.000000000","tz":0},"subject":"Fix IPV6","message":"Fix IPV6\n\nDepends-On: https://review.opendev.org/#/c/673291\nDepends-On: https://review.opendev.org/#/c/673777\nChange-Id: Id21850ce27f596d7bacc54fdccbaca766b40b568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5ec34722ee805024941d4e34634d1ebdf5275407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5ec34722ee805024941d4e34634d1ebdf5275407"}]},"branch":"refs/heads/master"},"9c7fa9048c51c81e7135429bd0f3d49b55b07263":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-08 01:24:04.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/51/673751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/51/673751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/51/673751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/51/673751/3"}}},"commit":{"parents":[{"commit":"7435a2e7bd7bb48b03f957a2edde32ad42e03fb7","subject":"Merge \"Bump to hacking 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7435a2e7bd7bb48b03f957a2edde32ad42e03fb7"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-31 07:55:23.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-08 01:23:59.000000000","tz":0},"subject":"Fix IPV6 rabbit host","message":"Fix IPV6 rabbit host\n\nChange-Id: Id21850ce27f596d7bacc54fdccbaca766b40b568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/9c7fa9048c51c81e7135429bd0f3d49b55b07263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/9c7fa9048c51c81e7135429bd0f3d49b55b07263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
