)]}'
{"id":"openstack%2Fmurano~676845","triplet_id":"openstack%2Fmurano~master~I8170d9c91f04e92a8374a403550bc05486733092","project":"openstack/murano","branch":"master","hashtags":[],"change_id":"I8170d9c91f04e92a8374a403550bc05486733092","subject":"Remove the deadcode","status":"MERGED","created":"2019-08-16 01:12:35.000000000","updated":"2019-08-16 04:19:21.000000000","submitted":"2019-08-16 04:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676845-1565929161991-674d1078","meta_rev_id":"d0eb37c883381e8006c23ea2cedb94c20340505e","_number":676845,"virtual_id_number":676845,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-08-16 04:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-08-16 03:25:33.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-08-16 03:25:33.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 04:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"655b60b4c54dd97f46b01e3eed855b0916459f30","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-16 01:12:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c679f1fc764d1771f4a92d431ef82ab3a371a4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 02:31:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e5fa51fff84e588254814096f6a5cf : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/839be764c4a442adb3e97777a4fa4a87 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f60119d575f4f62a30c218a2c4efe72 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/867ed199c32f4726bfbd990bf3e73680 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee62f6590f4543188c583b451c4558d2 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03465fadd33b4fb4aeef592dde64b9af : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99537db4efe4162bc040a043b75d9cc : SUCCESS in 7m 03s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/5c42420a7c524c818c4d830e834f79eb : SUCCESS in 36m 52s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/d413cd3432974451acb9c48667f95908 : SUCCESS in 35m 15s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/f25016dde3ad4c148511f198770c09b2 : SUCCESS in 41m 53s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/9b7da762f68842888dc1c6d91eea9d4e : SUCCESS in 36m 21s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/7b91fa1237844822baa474441f9854d1 : SUCCESS in 1h 12m 52s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6e9a49f5b4a14388aa5529c5525dcb4a : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4bf1edbe15b32cc72a5936ca9028f70ee45e3955","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-16 03:25:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64abfaa475b3cf6e30be27fe6c1b122711b7a784","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 03:25:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"677bdc9db2018ed411bb938aa7fcddf120be6993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 04:19:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d0eb37c883381e8006c23ea2cedb94c20340505e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 04:19:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44423dfd974e45f0a5bb0beb978fad0f : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea355f6fa35649bda7b19f5aab940142 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7e6c71ea3084d8fa7a0508b67ed250c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ecf83960bc148dd815b57cce838584b : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb2fa3a204ee4bfda0b161b00d08909b : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc6df6c23274b989f3770fa84445521 : SUCCESS in 5m 33s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/10865d69a4024d7480d1d126c6e61c88 : SUCCESS in 47m 27s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/21978e33977147858349e855932c137b : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b40e458c5f7c836e6db7a579b636ece915400779","revisions":{"b40e458c5f7c836e6db7a579b636ece915400779":{"kind":"REWORK","_number":1,"created":"2019-08-16 01:12:35.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/45/676845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/45/676845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/45/676845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/45/676845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/45/676845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/45/676845/1"}}},"commit":{"parents":[{"commit":"ae3f24f11091f0f9ee1ee54e1111a07a8e725034","subject":"Merge \"[train][goal] Define new \u0027murano-tempest-api-ipv6-only\u0027 job in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ae3f24f11091f0f9ee1ee54e1111a07a8e725034"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-15 11:32:01.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-15 11:32:01.000000000","tz":480},"subject":"Remove the deadcode","message":"Remove the deadcode\n\nChange-Id: I8170d9c91f04e92a8374a403550bc05486733092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
