)]}'
{"id":"openstack%2Fmurano~677423","triplet_id":"openstack%2Fmurano~master~I84791c93319dbd70fde436cc538b0e776e2a2a6d","project":"openstack/murano","branch":"master","hashtags":[],"change_id":"I84791c93319dbd70fde436cc538b0e776e2a2a6d","subject":"Add ip_version for NeutronNetwork subnet","status":"MERGED","created":"2019-08-20 07:54:15.000000000","updated":"2019-08-22 10:19:34.000000000","submitted":"2019-08-22 10:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677423-1566469019298-07838efc","meta_rev_id":"f7f9340a9b402e1cc759d6765a1122a47b6bb1ac","_number":677423,"virtual_id_number":677423,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-08-22 10:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-08-22 05:39:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-08-22 05:39:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 10:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"912c0a7d3fadc8f85f9b1a8fedddb8ca19975418","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-20 07:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51733d7682759c3cf8a23f403eafadf9934898a9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-20 07:56:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9b9fe9e29c7b2b7d01b00c4e089e48f7b273f38","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-20 07:59:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6e323f65d0d929ef458ca356a157248fe36afe3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:23:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d19382083ebd4774a3b1c2afdd79008e : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0cdfcb1b23c4c31a9fa2515f7bc04b2 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9419ab39b9943d886adfb369b5422ff : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6d08e75560d450a8a725af29671883f : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fe6638f63a246babc62caf1cd27a38b : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50059bac0f3f4a988f24e9a6fbc0a943 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80df262a11a74c3598bfebf2083c8319 : SUCCESS in 7m 13s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/ca933533e0a74cc083251a4fea7a7204 : SUCCESS in 45m 08s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/53c8c61c92a1486f84bf8ecd939d2931 : SUCCESS in 38m 23s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/1b64893a12d04f0497922297ebf29564 : SUCCESS in 51m 34s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/336d4c3038654f01bbe2efa6292cfb8d : SUCCESS in 34m 58s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/0770bb9453374f2b877bd2bc554dbefe : SUCCESS in 1h 15m 21s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d3d3102a551b420cb95c3a83a3e7b548 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e1cdf22a81d38ff15b92d9823f18a64bad14aea5","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-21 09:05:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"233f82ddc486399c417327eca569c52343e2e0d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 10:25:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ae2d0b8bb854f36acaa4f65d0e833cc : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d019e4614364bd7938db6b39019619c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3300c76dcc45929f22a0479dc26490 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b50d33a59a3436490b1671787e48d00 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d29c98743127436b8692707074a5ab8f : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa68e9a5b1204c858e7d92bc300d75d4 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a8d1db9c2447d48d9bf2377dea1930 : SUCCESS in 5m 28s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/69e668e1cb824d11ae9bc6a7fd7e753a : SUCCESS in 42m 11s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/000ccc49f7e4406d952d7f312258466b : SUCCESS in 39m 04s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/89f249f8a4b74f138cdeae4432519161 : SUCCESS in 1h 00m 49s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/cbb561cbdba24818b82f2cc16a191377 : SUCCESS in 36m 51s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/8b885872257c4e6c8af3066f2e9ee245 : SUCCESS in 1h 12m 49s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/803106b963fd488299c2f52b930bdddc : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":4},{"id":"184671635196d11ff673a290b5323059e8976f16","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-22 00:44:17.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a924da8bc754f08b40b627a3dc04fe28256bd936","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:03:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/778f2735be7446bcaeaa4fd378ecf40a : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/966f4a4a2a504631a0d91f22f5d40d4b : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c0614c296742b6952c2dceaf84bb36 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/963a9da45d144336b801c8ea078a7e40 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c380121bc44c42759c26e381303b41d9 : SUCCESS in 7m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4eacd38b8f1c403fa5bfbf2c7f73db44 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f01f6eccfb45ef8c6afb045dc0468f : SUCCESS in 6m 18s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/332eef66b49f46be903df362f6fee19f : SUCCESS in 40m 34s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/70104653a0ea4d69ac9c7e8c4e67b532 : SUCCESS in 39m 48s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/fe604d07c3bc4b3eb1c0c9a309caf70a : FAILURE in 56m 44s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/78be998e56ce42afa3bd12312ea0c3a7 : SUCCESS in 42m 13s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/5446bacf8f564d4cb496308396cee29a : SUCCESS in 1h 10m 04s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c5824f874e934270abee54da3b61c8d4 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":5},{"id":"e276024f45823ff202b61f0e139b01390a2f2416","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-22 02:43:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a696164b0d210c7a3b9974717ef69dbfbe0211b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:59:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3497ebb3d8db473d8eb70a32f1a36a64 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d3b8e1a1794cc8af51dc51409d612d : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51325b96c36d489ea6ee192639f7e7ed : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/985f140ab41b41b78645dfbc41d9278a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/602c43439626490e9640b89f792023d2 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffd4062866c942959c9f10cc44eca741 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ceeac485cc4e918652ce306a9d2ff4 : SUCCESS in 4m 51s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/bcaad8fccdc84ff29cc85b4dad0a35c8 : SUCCESS in 41m 29s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/5673ee362b7b44179afb5f65cb76e263 : SUCCESS in 36m 39s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/fa945fd07f524da890630a6777a1abf5 : SUCCESS in 42m 46s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/969561ea67c84b7184d72c30621c6dba : SUCCESS in 38m 21s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/3f4ba4e01ae547f1be543c32a0fe51a1 : SUCCESS in 1h 09m 22s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff9b95a2cb54e7fb546cdf3f6bd9a74 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":5},{"id":"8b4b403251ca86e83fd15233f2d7a24df27e8b05","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-22 05:39:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c625e323b82df50a8083dbc4ef2fc454b2feefdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:39:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"de6ec6ebc2c123b477e5c62e3a5ef44f232ed2cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 06:26:17.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91bfd835fc11444fae37b7064136bf0c : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859fc1b38d304bd3b237ccf098343cc6 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/837992cf339e439f94538a5609b08c6b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/288ecdb26cef4fc7a540b060b8e601a3 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1f8a115c40643cba4b33cebb115bfd6 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184a953a9f3b42eda7f29143e69724c0 : FAILURE in 4m 04s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/2d740dae791347afa556de26a4a1ff14 : SUCCESS in 42m 16s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/97116dbd6d7d440e92bcffc046e16b75 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":5},{"id":"33eeec57abf3de9beb5bc972c96d3617d558a8dd","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-22 07:45:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e4a7d4d823c20c03e9fd4fa6055db462852c2299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:24:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/029a7a58dc85455d840191522364479e : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5084fbb3649e4fc28cbd5a97271cc1ac : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e1c026e3ab4479acbaed2e80a85b77 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63774c80c3e04c96ba605045d20a359b : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7681e42eb8a841ca958f502a1098d395 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/617116a2f7a248f4924cc9a4a33535bc : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f46a45ec4144da991812fd83b6ade0f : SUCCESS in 7m 15s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/3de0996a42f04286a6e918e5859e8634 : SUCCESS in 43m 31s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/f3bfd0fc1f1b4b30813e58d8dc7ffe70 : SUCCESS in 36m 51s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/71179cc9e31c4eeea98dfa64581b8209 : SUCCESS in 45m 17s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/6c459180b1934b1892a5a14401dbb2c0 : SUCCESS in 40m 36s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/cc9364ee289847ecae279738162c5b7d : SUCCESS in 1h 32m 10s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d307cce83d77417bbe13d11e537210df : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":5},{"id":"e362d43955f705738641111427f977375a1bea9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:24:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4d65b32c846588021678c17da21a9c5d4ec06743","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:16:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e401c1f33862b48587c86d53ebcd0a3570c797d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:16:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4ba3d3f5b594634a17e70b556c62c01 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a8e2b398c1472f81d5e1f1ef7c833a : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f231d0e533fb4a33bfe6a8e55fb9f90f : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99a1bdea8223497ea09454c5e2ec6f87 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dba293dc3282480aadb4425a20953f8d : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb95f5abea64b2e854b5353c7dbf6db : SUCCESS in 5m 29s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/840cc9f7a41f4b568bb48cf33325364c : SUCCESS in 46m 44s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb154871c41491aadc712be80fbdf5b : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":5},{"id":"f7f9340a9b402e1cc759d6765a1122a47b6bb1ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:19:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836962a679df4400a2c96d54abf56a1d : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"81acfe0f17a4e9f3a0ea4bda994aef6091132c53","revisions":{"b88bfc83c0bd9bf27817b4a88e07be9a49cf87b3":{"kind":"REWORK","_number":1,"created":"2019-08-20 07:54:15.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/23/677423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/23/677423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/23/677423/1"}}},"commit":{"parents":[{"commit":"b40e458c5f7c836e6db7a579b636ece915400779","subject":"Remove the deadcode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"subject":"Add ip_version for NeutronNetwork subnet","message":"Add ip_version for NeutronNetwork subnet\n\nDepends-On: https://review.opendev.org/#/c/676900\nChange-Id: I84791c93319dbd70fde436cc538b0e776e2a2a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b88bfc83c0bd9bf27817b4a88e07be9a49cf87b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b88bfc83c0bd9bf27817b4a88e07be9a49cf87b3"}]},"branch":"refs/heads/master"},"4b0532ee3cf42d82dff17fba473a21cbd1fc5e15":{"kind":"REWORK","_number":2,"created":"2019-08-20 07:56:46.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/23/677423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/23/677423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/23/677423/2"}}},"commit":{"parents":[{"commit":"b40e458c5f7c836e6db7a579b636ece915400779","subject":"Remove the deadcode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:56:27.000000000","tz":480},"subject":"Add ip_version for NeutronNetwork subnet","message":"Add ip_version for NeutronNetwork subnet\n\nDepends-On: https://review.opendev.org/#/c/676900\nChange-Id: I84791c93319dbd70fde436cc538b0e776e2a2a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4b0532ee3cf42d82dff17fba473a21cbd1fc5e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4b0532ee3cf42d82dff17fba473a21cbd1fc5e15"}]},"branch":"refs/heads/master"},"55473daa0444b6482ffb643d24e0b17f4e77df65":{"kind":"REWORK","_number":3,"created":"2019-08-20 07:59:40.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/23/677423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/23/677423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/23/677423/3"}}},"commit":{"parents":[{"commit":"b40e458c5f7c836e6db7a579b636ece915400779","subject":"Remove the deadcode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:59:20.000000000","tz":480},"subject":"Add ip_version for NeutronNetwork subnet","message":"Add ip_version for NeutronNetwork subnet\n\nDepends-On: https://review.opendev.org/#/c/676900\nChange-Id: I84791c93319dbd70fde436cc538b0e776e2a2a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/55473daa0444b6482ffb643d24e0b17f4e77df65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/55473daa0444b6482ffb643d24e0b17f4e77df65"}]},"branch":"refs/heads/master"},"37d1223ef44f6065971b9e4c112ca56ce9c3e11f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-21 09:05:15.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/23/677423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/23/677423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/23/677423/4"}}},"commit":{"parents":[{"commit":"b40e458c5f7c836e6db7a579b636ece915400779","subject":"Remove the deadcode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-21 09:05:11.000000000","tz":0},"subject":"Add ip_version for NeutronNetwork subnet","message":"Add ip_version for NeutronNetwork subnet\n\nChange-Id: I84791c93319dbd70fde436cc538b0e776e2a2a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/37d1223ef44f6065971b9e4c112ca56ce9c3e11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/37d1223ef44f6065971b9e4c112ca56ce9c3e11f"}]},"branch":"refs/heads/master"},"81acfe0f17a4e9f3a0ea4bda994aef6091132c53":{"kind":"REWORK","_number":5,"created":"2019-08-22 00:44:17.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/23/677423/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/23/677423/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/23/677423/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/23/677423/5"}}},"commit":{"parents":[{"commit":"b40e458c5f7c836e6db7a579b636ece915400779","subject":"Remove the deadcode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b40e458c5f7c836e6db7a579b636ece915400779"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-20 07:52:43.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-22 00:44:12.000000000","tz":0},"subject":"Add ip_version for NeutronNetwork subnet","message":"Add ip_version for NeutronNetwork subnet\n\nChange-Id: I84791c93319dbd70fde436cc538b0e776e2a2a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/81acfe0f17a4e9f3a0ea4bda994aef6091132c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/81acfe0f17a4e9f3a0ea4bda994aef6091132c53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
