)]}'
{"id":"openstack%2Fmurano~694646","triplet_id":"openstack%2Fmurano~master~Iac87c01defbd7de3f14f13e5632aec899d920d1b","project":"openstack/murano","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Iac87c01defbd7de3f14f13e5632aec899d920d1b","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-16 00:19:33.000000000","updated":"2019-11-26 01:45:48.000000000","submitted":"2019-11-26 01:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694646-1574732579168-a4121484","meta_rev_id":"0c69c001c0c4ff16ac74c805c60dc24650cbeee4","_number":694646,"virtual_id_number":694646,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 01:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-25 15:31:54.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2019-11-25 16:59:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-11-26 00:57:11.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-11-26 00:57:11.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 16:59:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-26 00:57:11.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-11-26 01:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"859c310e8844e5051afe10ef45a42b7441aae187","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-16 00:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"083b07fa88071a6e32877f522fec520a34d2a2a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 01:44:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f446b86cb3664ad293fa49176755345d : SUCCESS in 4m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df2bc4c0d2d244fa9d0e6f96fba6d1cf : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6e54435050f46f080f50078ab9a4e79 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d3e44741324823a2771225a1677753 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e283c4689ec4016b0c22a7aac04f891 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a7780b4e074f75a3b2e19899a62628 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc5d1c5c48446d7ba4c70e9d1265751 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2950ce718f604f7f89a6cf15e40704f6 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d291ced02aa45539a303770e9282b98 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be9fb4f30c7245e1886017f2f6907908 : SUCCESS in 6m 45s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/3823080ed1e740019f9377ef6b2955e0 : SUCCESS in 36m 54s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/1ab0b06733674da59f66e7534958ab05 : SUCCESS in 35m 07s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/c5e9466928454034ad8035915d77fa67 : SUCCESS in 44m 17s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/12d517ff28604cdd81f859ad487666ae : SUCCESS in 29m 48s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/73d36ed48f8443e9ab1b28143bf078b3 : FAILURE in 1h 22m 29s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8b686364e3724af483a593370ba94634 : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8355bd649a12d82d01a559f2c6fb07518c0cfd73","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-16 14:00:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d566e7e0c7b7e36cc5f8011851d8b3d8d952d367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 15:27:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1c0845ba47c49b1a5fc88bbe930cc99 : SUCCESS in 5m 42s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/103af3a93bb340fdad6cc05030a131c7 : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b6c869166f94c4991f52b7a90e4d1aa : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e2020268e5043f29cf26449011b5304 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/704f58aacb0b4aea8af8a50da3df230f : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ab466bb62e4091a4a639c48be34a9d : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d1d570285d4c1e9520354084f35fd0 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e1da471e1f246f281c24377ffe5eafa : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41d9c2deaa74ccbad1d6f098c6b5c73 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6436e1fd73e545d3a33968c4b6ed7ecf : SUCCESS in 7m 07s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/37b745fdac024b7c801345175afd513f : SUCCESS in 38m 12s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/ae86991b4bd5435aac5f4cd9ea8e82ae : SUCCESS in 33m 40s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/3e6117d07c2849c7a7dc5c59982ea363 : SUCCESS in 38m 27s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/12da11e7302748348e95203ba464bc57 : SUCCESS in 28m 36s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/c83930ef97e04463b16cb0036de718c1 : FAILURE in 1h 25m 43s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cc98f7f11023490d9a1c416dd16a4f17 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":2},{"id":"46463213e878ee3292e8f50e256875535c5ec300","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-24 12:42:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d57ffb6eae1c7d8ca1af96783c08e5670e1e5b9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-24 12:47:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee1a0468d178c49d6b99935060b6ffc32329beee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-24 13:42:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03ed5019b25a4a6aa149462eecc5285d : SUCCESS in 5m 37s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/051895c432cd465ca4eed661f1d62a14 : FAILURE in 3m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16c29f439a4344efb3b6f77c97398816 : FAILURE in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c154597cab5427b86ee8fec5d7d8c31 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b88d55c2e44f4500a0f88a2cf6dec102 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e236d3eb825344f784112201f3a80d58 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2bc48da30f481da4b5f54b4a9438c3 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf8034335c24dd190da5384371e86d8 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb83cc460244f62ba559bd9af4cfbd0 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74ef4f92100470e9048aec1c86b08d6 : SUCCESS in 6m 16s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/24bfa91d2497461b856cb5b043ea693c : SUCCESS in 38m 14s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/c4f705a0cb574c728082791cbb9f077d : SUCCESS in 36m 51s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/2bfd524472624d72bbe9f26eb006308d : SUCCESS in 53m 35s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/f617d2bf2d7b46868b1c35310e8bc5a3 : SUCCESS in 36m 45s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/baf3b348a3e448ad8fb516a8b161cda7 : FAILURE in 16m 03s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a962bb90faf748289efe2f5a2aa142e1 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":2},{"id":"43fc3205f5fdd20f78abbc49c4fd7eca5fd85d87","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-24 16:18:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a017386cc8564c65ec17f4b8f7094ddc10f192b9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-24 16:28:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"df7680220b65278c7ae2835311bc6e2cc889d066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-24 17:16:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30f385429e3f45bf981ded2046f8766c : SUCCESS in 4m 11s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a68d7094fbce45b0b9ccdaf16d643008 : FAILURE in 2m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/35e77e950f1e4f02a15c38cff94f0b1e : FAILURE in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed624fbe709f4a9f8a222ab776a0aca0 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23c87bc795af49a8a4fbee2b66b30e27 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34bf5b626234ce494fd609214e6f7b7 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da940b5cf61d4bb7addb83faa78f295d : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edd3696b82514b52847ebb6992a0032e : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501f02bd43ff4e5b9106a0e8fdb0dd4f : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa1b5f2a6f3a42ac897ceba7ac3bb0b6 : SUCCESS in 6m 41s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/94268c0c370e468ab3a50aef85568e47 : SUCCESS in 36m 05s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/32084737ac334a4ba20613a0a4619aaf : SUCCESS in 33m 48s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/a9c634f7b5e34d20814dca63fd49cf51 : SUCCESS in 43m 01s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/c55d212ac90c4ee8836df5f6cb776c72 : SUCCESS in 32m 54s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/e565af39925c4f45b50e1b9084908a6d : FAILURE in 7m 37s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0748cfcd0fb44ad8a582e25a485f814c : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":4},{"id":"40dce4b6954043d532b0856ee317e8ca5c1e1032","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-25 07:54:31.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b8978e090dead1f23ed7d53455ab0726f3dcad72","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-25 08:28:15.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"8c2727735c16bfe7492b38b3164b16e9b3ff83bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:47:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/51674c1afb9c46d6930c88e3d7ecf7aa : SUCCESS in 5m 48s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b8579b143224182858634268c719bd7 : SUCCESS in 3m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d497eaed6d664d488ceec2d2c2d95ccc : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b305213820494ae18ad6587580ab482c : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de5e73d5fdd74b2c82bc726be866b2b0 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cef6ed749e647ee8dbc4beb0dee85e9 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1ce9ea0323466d81fdecf1f4c9bc83 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbead0f94eb14cd795727ce135f20460 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d653b2cdd3b54e1d9db2b33d92f9c194 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08366b262e1041fdb092c80e8106b969 : SUCCESS in 6m 50s\n- murano-congress-devstack https://zuul.opendev.org/t/openstack/build/d1f4fc7c0f57472bab259add2650bf11 : SUCCESS in 39m 12s (non-voting)\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/ad5a31d648764c0da91d3b1a6008b29a : SUCCESS in 37m 13s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/fb8bce53dcfd4360af79ac84f025a047 : SUCCESS in 54m 29s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/1d8e180cfc7a4495bcd95bf2f61a60e8 : SUCCESS in 35m 39s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/f520211d49c1449ba65c1f31c2d5a1aa : SUCCESS in 1h 13m 35s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/88f460a44a8445a99f9b45c606edf9f7 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":6},{"id":"cf7cddc93a29166793a52bdd8112916a41ea890e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-25 15:31:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a3729280f33f3c72c7620a42cf774760c9942658","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-25 16:59:16.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f711e531c1b78d77720932193e1f9096b4b7a4df","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-11-26 00:57:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"caabb0f20e5ff8a675c8f98bc76e36cfd694f66b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 00:57:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"368aaa2776a05e4b76ea5a001211ebf0c7de0fca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 01:42:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ee36775801c47d49ec908421d57fd01 : SUCCESS in 4m 08s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d507fd643a64f90b481c0e82808a878 : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f9518d7db2e493ea7b7d666b27948d9 : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6737f098627c46878a8314a9d6c294ab : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb3dc6d1ba040a7bee72aeb08f6daa5 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df95f9d64da4b7597d46311ea97a20b : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3400486bbc44aef8c5c5dfb5276366f : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75f4e8fb3174a448eb20361534bfb6e : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4d0410567346cb895a17fa61db62fa : SUCCESS in 7m 15s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/ccf82d32fced47c18025cde9b9ad0ca9 : SUCCESS in 41m 27s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/70b9d39c830340beb77978dfdb52f55a : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":6},{"id":"d40e51f0a7cf3dfec64abafcbed9fa12827a844f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 01:42:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f26f332ee1dbdb7013c54312bbaefd4d84f1971b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 01:45:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e51cf4ce88ef4bfe8d8cdd38959e2501 : SUCCESS in 51s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc35cf192de4400b1b126e07b0bd000 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e5f451baf654c7d9c93084904cc6fde : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5e12408b76295b5b2abf3d1424975d3a5216a1c5","revisions":{"ee059f086dd288032e6b5a04116bc1af4dad3f86":{"kind":"REWORK","_number":1,"created":"2019-11-16 00:19:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/694646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/1"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:19:31.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ee059f086dd288032e6b5a04116bc1af4dad3f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ee059f086dd288032e6b5a04116bc1af4dad3f86"}]},"branch":"refs/heads/master"},"36ecc40eb0b68c81525c7ce54ae26f160b7d350a":{"kind":"REWORK","_number":2,"created":"2019-11-16 14:00:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/694646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/2"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 14:00:05.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/36ecc40eb0b68c81525c7ce54ae26f160b7d350a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/36ecc40eb0b68c81525c7ce54ae26f160b7d350a"}]},"branch":"refs/heads/master"},"5b097af468b1fa697f1ba39da0fd8dd115c05936":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-11-24 16:18:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/694646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/3"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-24 16:18:32.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nDepends-on: https://review.opendev.org/695821\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5b097af468b1fa697f1ba39da0fd8dd115c05936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5b097af468b1fa697f1ba39da0fd8dd115c05936"}]},"branch":"refs/heads/master"},"ad6af07b5d7ba915b8fa852a14bf3420dc73a22f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-11-24 16:28:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/694646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/4"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-24 16:28:13.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nDepends-on: https://review.opendev.org/695822\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ad6af07b5d7ba915b8fa852a14bf3420dc73a22f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/ad6af07b5d7ba915b8fa852a14bf3420dc73a22f"}]},"branch":"refs/heads/master"},"f073646e46fcd121232fe749d2fdb53ef58bcae5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-25 07:54:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/694646/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/5"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-25 07:54:22.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f073646e46fcd121232fe749d2fdb53ef58bcae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f073646e46fcd121232fe749d2fdb53ef58bcae5"}]},"branch":"refs/heads/master"},"5e12408b76295b5b2abf3d1424975d3a5216a1c5":{"kind":"REWORK","_number":6,"created":"2019-11-25 08:28:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/694646/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/46/694646/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/46/694646/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/46/694646/6"}}},"commit":{"parents":[{"commit":"d536a70ab40a16649301b6710ed14d05e863ae47","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d536a70ab40a16649301b6710ed14d05e863ae47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 00:17:25.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-25 08:28:00.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMurano is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: Iac87c01defbd7de3f14f13e5632aec899d920d1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5e12408b76295b5b2abf3d1424975d3a5216a1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/5e12408b76295b5b2abf3d1424975d3a5216a1c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
