)]}'
{"id":"openstack%2Fmurano~745481","triplet_id":"openstack%2Fmurano~stable%2Fussuri~I64d205b406ac220e481fd8d3c1d611a7b24f32f6","project":"openstack/murano","branch":"stable/ussuri","topic":"native-zuulv3-migration-stable/ussuri","hashtags":[],"change_id":"I64d205b406ac220e481fd8d3c1d611a7b24f32f6","subject":"Add native grenade zuul v3 jobs","status":"MERGED","created":"2020-08-10 06:33:20.000000000","updated":"2020-08-11 09:25:54.000000000","submitted":"2020-08-11 09:23:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745481-1597137784045-ea02c699","meta_rev_id":"bced4d71b323aa5986eb94dcda8aa8861820c510","_number":745481,"virtual_id_number":745481,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 09:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-08-11 08:19:46.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-08-11 08:19:46.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 09:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7a844d0b44d441d8ea3fa9b272b57b919f51377","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-08-10 06:33:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8821af019813187479a4329dc90db4f1cff72eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 07:43:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6b7989f68d47919a6cf853349a4ea3 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04bf09ed52eb4b5899efd3250b54b0b4 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66117fcd59ce42d6809ffc5c4c9c278a : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8576a7f02b4675a1301da58714b329 : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5471a4c32da477da500e4e270ab738a : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de0be0804674d2e9db2e7d7e487862b : SUCCESS in 8m 24s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/f9e193a773e24177b59b8ba3e8840ce2 : SUCCESS in 40m 28s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/360bcedc9e4a40c1a2560ce9774da2e8 : SUCCESS in 47m 19s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/b7839d9a40f84d2d8a6342740f3d2b47 : FAILURE in 43m 43s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/9bbdedc1af214ea0b1d99b602c245977 : SUCCESS in 1h 02m 18s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ec068d95ca40978d36a37a1ee8af11 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0eda70b4257363d0167e12c10d3e8726514f3118","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-08-11 08:19:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c84b551132e14133ca3b04daa852efaf58346bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:20:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4009646252acb92ac08dd86105210c245591a1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:23:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea66fd865e85404193cc7662d74535bf : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a746eb264771441fa366d3f855e3e814 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/debe3e2fb8a24bb5b61aa50fe1a251e5 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4402f3647688414781ae1472c698b9b2 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c687485071064c65a8b4b79844d53a40 : SUCCESS in 9m 40s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/b28fa4d1105c4d79b731b9386bff0edf : SUCCESS in 47m 07s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/032bb6bacc434721a1d587c8f138cd78 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fdf90c96c120446bdd6da4fcc910c308ef66d3d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:23:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1bbcfca9853a0f6fbf980118d7fbf54ed6bad606","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:25:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03160e4102044c20bb88fc8f7bed2b15 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1daccfced80ae3dfac59f08fd20d31244ef9d94a","revisions":{"1daccfced80ae3dfac59f08fd20d31244ef9d94a":{"kind":"REWORK","_number":1,"created":"2020-08-10 06:33:20.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/81/745481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/81/745481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/81/745481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/81/745481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/81/745481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/81/745481/1"}}},"commit":{"parents":[{"commit":"29c5ef182a18443dc2a717caed06e6695463288f","subject":"Fix murano ci UT error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/29c5ef182a18443dc2a717caed06e6695463288f"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-05-14 10:56:09.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-08-10 06:33:20.000000000","tz":0},"subject":"Add native grenade zuul v3 jobs","message":"Add native grenade zuul v3 jobs\n\nDepends-On: https://review.opendev.org/#/c/731681/\nChange-Id: I64d205b406ac220e481fd8d3c1d611a7b24f32f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/1daccfced80ae3dfac59f08fd20d31244ef9d94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/1daccfced80ae3dfac59f08fd20d31244ef9d94a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
