)]}'
{"id":"openstack%2Fmurano~768520","triplet_id":"openstack%2Fmurano~master~I1b6c6485bc651fd0b87244a68204036dd4aa37f4","project":"openstack/murano","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I1b6c6485bc651fd0b87244a68204036dd4aa37f4","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-12-25 21:47:29.000000000","updated":"2021-03-08 07:08:18.000000000","submitted":"2021-03-08 07:06:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768520-policy-json-to-yaml","meta_rev_id":"c2bcdb525be95ca22bf271d416ac8e636774fabb","_number":768520,"virtual_id_number":768520,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 07:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-03-08 06:07:09.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-03-08 06:07:09.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-25 22:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 06:07:09.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"d315dfbcb0de2c18c829c1be8804da6427cca920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 21:47:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae0a4aee317917b72ff9851b6c0c87c53f56ba5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 22:46:45.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb88f978935048eab197bd84ae3afe47 : SUCCESS in 2m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb0fe72ff981455897f97b6fe2579e76 : FAILURE in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cc0cfd0ee0b4708b0906663b0ce1f81 : FAILURE in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72a49fc2a6c34464b999934067c904d4 : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4100b9dd314452a6b98334fb11879b : FAILURE in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f0bafc51a14d1589f13d58385c2067 : FAILURE in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f369eb8a38d54d3d9a7fae60d1d0821a : FAILURE in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da80783be6144d3abff063976a163b27 : FAILURE in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d48eba6c9fba49d8b653dd3ec90225aa : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6a1a2275cdc4d199503af7373cee926 : SUCCESS in 5m 54s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/7d6fd13315b744b2aa2663c8711c01f6 : SUCCESS in 38m 21s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/44a859600d2f482684a5d3e9badf0e07 : SUCCESS in 51m 10s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/6ff506103f464a80a3305fcc4f63778e : FAILURE in 39m 11s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/01bd8bcc1f5648879d80eb4ce0db9796 : SUCCESS in 56m 27s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7578ac75f7e345ebb50a8cc5f878ca26 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6b8cfea5e1d38cf59733703469e490f9e074181c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:14:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"043067976a2bc03eb0a0c4e6f79500525b8c5745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 21:16:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fef0d62934c849fa946bbe40a7ebb55a : SUCCESS in 1m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/89c520cdb718428f99dea744900ec7a4 : FAILURE in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec6f77797fc46eca51b58035d85ef87 : FAILURE in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf23378a561468db75f6448425616fc : FAILURE in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66284af0801247f5aafb17d2d989f618 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb270a02f074b85abc4393db3df3615 : FAILURE in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbd4515682d46d1b91231223a16c2ee : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de13ec0f69234abdbc2cb7b632033f3c : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b5f86838e340cf9b683063d45c53d3 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bccf8b650e94facbe0ba60acc513a23 : SUCCESS in 4m 32s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/558323b7dbd04b5f85a90cb0dc1c7922 : SUCCESS in 39m 39s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/c625a3c70a8348d9a0bef655e8befb73 : SUCCESS in 48m 09s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/b175c5f715804df7a4c32bd7823099f1 : FAILURE in 40m 13s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/16360db126724a008fecf794cd857d78 : SUCCESS in 56m 07s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/474eb934824649a88a974f5fd317b9b8 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c8b98846597c10d1e98bcf72ba2347888baecdc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-29 22:36:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"570377b5a8c65b3da17f9ef32c4a45846fea461d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 23:34:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5146f2a289494aa19571a66100c0c251 : SUCCESS in 5m 04s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5e9de54b92944899b6fbec677e61c36 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/374a09e6b828469ca41dbcad2e51b0d9 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/118b053ecba342d393c3259485f53c36 : FAILURE in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5ddb73ea45a4035901195531fefccec : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d15ce6af17443ca1ee293ede3a9399 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aedaf242ddc412b90906a00a79c2529 : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f0062da7cea4d7ea4ba813586281d21 : FAILURE in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aab7a16f0bc741bfb4a5ef4563b4d3bc : FAILURE in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3115f0f28f7d425ebff00d42fb292a54 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48f9f054dd7f42b18aeda9fe192c866f : SUCCESS in 3m 59s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/231a1e10a4fb496f9a1fbeea51e63710 : SUCCESS in 35m 36s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/07d073e66faf4f1992f65b13e9b474c0 : SUCCESS in 41m 49s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/7825d9e8bc60473eabf5c4a06d248145 : FAILURE in 49m 14s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/42d3873f8e9d4ba084441dee982d2b47 : SUCCESS in 46m 06s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f5e73e81b15b4e4a81ce8ae843da6570 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fbe837f435d73ac1c346be2ae0090c383bdcf5f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-11 18:46:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6880890868d75c6c0a8e8ae086e56a6f58b28957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 21:24:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93d76543fa994615aa7e5d6dfe3cbb7b : SUCCESS in 6m 30s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f72ca43faa0145c29339ce4a45abd37d : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4f05aa0ada041dcb294fb639ba9d115 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdaf1c6db57e41c1b0ed415dadd531d2 : FAILURE in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b8a41cd7d2c4e77850cdf21d5c2982e : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5a2612704445fb9e817b2bde4cfd2f : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47605da2fbf34213835e53b3444de4e1 : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17c48c3445e04e6c961aceeee36e9823 : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80a5b1705e344e9dab9e2e8f4e0d4714 : FAILURE in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500b7515e5b54aa1b50fc17e962e9286 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19991538d0d4ad68ac4e3402ccd4ed1 : SUCCESS in 4m 49s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/20c1d2b5196142e0a5d273b35a914edf : FAILURE in 2h 13m 13s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/4cb69245fbc045be9fb43cbc326df744 : SUCCESS in 44m 21s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/778d90ef4e1c4431b3678cdcbf26732e : FAILURE in 37m 56s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/4a3f3585cc5b48488e793eb6d47bc973 : SUCCESS in 49m 00s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cc24f923e5cd4d15ab232be64f761239 : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":4},{"id":"f599a9daaa02e2e86c307d105f9f98bf7d67fa6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-12 17:23:25.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"43fa1969279b1857201d8adbc366b3799d2c2e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 18:30:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/747491e8be984a45b5e148a6cdc69dbd : SUCCESS in 4m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2766a636291f40a98a084f1a310931d7 : SUCCESS in 2m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c30a42bc750c4003b75f544c22843db8 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48f93765ce964952941af26e96e2b15f : FAILURE in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e559f3c63c0420eb37c56ff282855d3 : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9724b61fb708492391c20299c595dc46 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02d5f5c2e540455484e7bfd0384f3a94 : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79643541f5e045f18d0c6e3969200c57 : FAILURE in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/585010539eae4545a97fe73eb613f206 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55682323b88947d49551765d97e4bc1e : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47ffe88960574be18b5404b45fd424ec : SUCCESS in 8m 19s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/15da8f1b84ed46be8c4b872ee3988947 : SUCCESS in 33m 24s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/bb5f43540aa34a65ba7cefd88a5155e6 : SUCCESS in 52m 16s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/b3058098bb5c408db3db1871621ac184 : FAILURE in 39m 38s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/e2aa88416f8d436899df796c2980b5aa : SUCCESS in 54m 10s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/106d435d8d58421a9db5eb322dac5fa1 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":5},{"id":"c11d1663176d7884735650c8892d26609ffeefe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-17 00:44:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ed71253048c4e3627e5de48e202811b7f1f0a026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 01:30:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/63179231cc704b5fbd23c843bed4fe87 : SUCCESS in 5m 37s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d24ccbc72a034434930bf14a88db64b5 : SUCCESS in 2m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4283f36430e849c9881d606e3d2de918 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e08be6b6c7004895b0921434f2ca6d87 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e08a48ab048248c294dea6ff771dd6b3 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d664115e614827a25acbe4d6d936c7 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/912fc6c5d5bf44aab43a2a7c64b88f05 : FAILURE in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c26abd1c334418f95de5f7f8f47edaf : FAILURE in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95b3a0608e1c417eaf61e2dfaac26e8d : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9d032093f847a5b4ed5ac06a09f0fa : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712dea64571a4ba9898f2b82889355a1 : SUCCESS in 5m 06s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/d0e10783b47a45b8ac65191fbbb2b012 : SUCCESS in 33m 27s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/be94501d412848b195322bd790400ef6 : SUCCESS in 43m 07s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/506875d495414447b02311dcf77c343f : FAILURE in 33m 11s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/7f68943fb84947e5bb2136b7fbc9d036 : SUCCESS in 39m 46s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d4f116841ff04ef0907a07e00035a359 : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":6},{"id":"95e03e3cda7bfe15f63c3b04fdd5f95d4d800164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-07 01:43:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00523d5a2537867b0b911ce943b402b7bd1a898e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 02:49:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21f2580484aa4c20b5d1cc67d4ac05c3 : SUCCESS in 5m 01s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/814ecd9900bf4289a1030bcdf3b61814 : SUCCESS in 2m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/115701ab8a34481a9455c83a0496d538 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5581ca365c69437b9f3ffe8e10a25ccd : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dcb6122b7174b7997b5d7d8fb230110 : FAILURE in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247b0ee4e8094f0d8cc3aeb09979ccd5 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/353a18911573444ba3285fb24de4f256 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b0cf5853b046bca146b7647284abc9 : FAILURE in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123385bd796b464eade06892f3795cc2 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcb5160f45f43dcbdf9db3f993ced84 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/742d9f13f9fb4ea7ad1eccbc7e562a32 : SUCCESS in 4m 51s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/947b6765168e4bb980f520b7364220b6 : SUCCESS in 45m 22s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/fc2b5a5be3ae49698d328f3bce7219cc : SUCCESS in 41m 46s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/7e9c004f4e9446878c54f9f7e13f7b0e : FAILURE in 28m 44s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/32567596d84446c4997763237a17195f : SUCCESS in 1h 00m 31s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/3550be52942d420dbbd271b91bc480ee : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":7},{"id":"82463dad2cce32827b43061b90f7a043ee8f4b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-07 23:50:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d76742326feb4eb4990e23c8a99f295f22486d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 00:44:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ede4414e0c04adda41ba854fb280659 : SUCCESS in 4m 58s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3486f56b7d2845118ee09657b2f9436b : SUCCESS in 2m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b73b865c43d41f79857b171bca97dd5 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7484a3f6b34c45acb8a24b5523d6d6a0 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/907edbacbf0a47cc958372cf50d2d46f : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a47d2c50244d15a7105bd5d7d28d7f : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ae7ad4893d4c9b9f0d2c0b231a7a34 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e40b005ac8e4972aff92b5a4bdab258 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/484048941da245129f8ffd95e18ab945 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7316a8ccf39e4c3bba1e5c7f13933585 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5d5916050b84dc6a961a75f72eab29c : SUCCESS in 4m 39s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/f8611e20623f47d3b05ec2cce92feda6 : SUCCESS in 36m 09s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/af9429595e234a25a38fc87002006cf5 : SUCCESS in 44m 39s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/aa5798b476e84f349eb4a4d41a256c34 : FAILURE in 37m 28s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/1714dbe571944b46b631255da15e95cd : SUCCESS in 52m 27s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a689781f437b4035831cb47b62edc5de : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":8},{"id":"8a153b5323f2f213feca69d353c289fee9d9b402","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-03-08 06:07:09.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f46b9549533861f53bca5b6dafc4c1bb4cc1cb82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 06:07:22.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"32ebaf24335f5cd0bd1176e80915816569d0d902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:06:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2cb82a9ce579429192f48bc8f18d3914 : SUCCESS in 6m 57s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23fe55551cf842ca858f7a4056a00063 : SUCCESS in 4m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84b24cff26a842fa838ea713fc2f5db5 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f93e0aad63b40b483fb9093e0bc0b4b : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f290f22629204a80b82d2454266b947e : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b47e4d8b26f5463cb65b5d30d823f874 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ad9cc3193de44d885a9792f38b24951 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b885ce935324212bceadb25ef2d30df : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e23107a7e8e42c1b1c8ed9de4c7c310 : SUCCESS in 7m 57s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/8c2405bded6245b9bd7ea4d80de20ec1 : SUCCESS in 40m 38s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/37aca403e66c4a2b9d9bc601b97db4b3 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":8},{"id":"9ccd2bc0c1851ad759a87ae33f046f453c047036","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:06:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"c2bcdb525be95ca22bf271d416ac8e636774fabb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:08:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2075e46a7c0a4f1f87a2cb192a527b30 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd18dd00b9a4781ab6a06fb59b0bb9b : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af8a160b47f14c47ae7c795371449bda : SUCCESS in 51s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1c87ebf9a8a79d990e4a1ba44452b50efae77883","revisions":{"7032b4a4323d11c4e50cc69588a4221d9373a147":{"kind":"REWORK","_number":1,"created":"2020-12-25 21:47:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/1"}}},"commit":{"parents":[{"commit":"e078063cffeab325d98025c43b7a955c53dfc78c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e078063cffeab325d98025c43b7a955c53dfc78c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:47:28.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7032b4a4323d11c4e50cc69588a4221d9373a147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7032b4a4323d11c4e50cc69588a4221d9373a147"}]},"branch":"refs/heads/master"},"4b42a54465819b2c511f5d66d83f932a60789492":{"kind":"REWORK","_number":2,"created":"2020-12-26 20:14:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/2"}}},"commit":{"parents":[{"commit":"e078063cffeab325d98025c43b7a955c53dfc78c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e078063cffeab325d98025c43b7a955c53dfc78c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-26 20:14:42.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4b42a54465819b2c511f5d66d83f932a60789492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4b42a54465819b2c511f5d66d83f932a60789492"}]},"branch":"refs/heads/master"},"260e75d3cb476fe5fa38d63488a262e472805a64":{"kind":"REWORK","_number":3,"created":"2020-12-29 22:36:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/3"}}},"commit":{"parents":[{"commit":"e078063cffeab325d98025c43b7a955c53dfc78c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e078063cffeab325d98025c43b7a955c53dfc78c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-29 22:36:35.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/260e75d3cb476fe5fa38d63488a262e472805a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/260e75d3cb476fe5fa38d63488a262e472805a64"}]},"branch":"refs/heads/master"},"32c55079ba1f4dce4f5850012f06af0a4dd5fbaa":{"kind":"REWORK","_number":4,"created":"2021-02-11 18:46:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/4"}}},"commit":{"parents":[{"commit":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","subject":"Use common rpc pattern for all services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-02-11 18:45:58.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/32c55079ba1f4dce4f5850012f06af0a4dd5fbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/32c55079ba1f4dce4f5850012f06af0a4dd5fbaa"}]},"branch":"refs/heads/master"},"16bd84b2f99a266eac1a6f4cd44e2a2e5fa30e33":{"kind":"REWORK","_number":5,"created":"2021-02-12 17:23:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/5"}}},"commit":{"parents":[{"commit":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","subject":"Use common rpc pattern for all services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-02-12 17:23:21.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/16bd84b2f99a266eac1a6f4cd44e2a2e5fa30e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/16bd84b2f99a266eac1a6f4cd44e2a2e5fa30e33"}]},"branch":"refs/heads/master"},"87bc018be16eee523424c712482fa43dc28f1fbd":{"kind":"REWORK","_number":6,"created":"2021-02-17 00:44:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/6"}}},"commit":{"parents":[{"commit":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","subject":"Use common rpc pattern for all services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-02-17 00:43:42.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\nCONF object needs to be initialized before policy enforcer(). That\nneed to remove cfg.CONF.unregister_opts from TestAuthUtils cleanup\nas this is taken care by cfg.clear() with proper workflow otherwise\nit end up with error\n\"oslo_config.cfg.ArgsAlreadyParsedError: arguments\nalready parsed: reset before unregistering options\"\n\n- https://b132754ee7062a9ab187-9add4719a9922a9385555a8552fc2366.ssl.cf5.rackcdn.com/768520/5/check/openstack-tox-py38/7964354/testr_results.html\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/87bc018be16eee523424c712482fa43dc28f1fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/87bc018be16eee523424c712482fa43dc28f1fbd"}]},"branch":"refs/heads/master"},"45102b913db0dbcbe97d24392bcebf45ce5f2697":{"kind":"REWORK","_number":7,"created":"2021-03-07 01:43:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/7"}}},"commit":{"parents":[{"commit":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","subject":"Use common rpc pattern for all services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-03-07 01:43:19.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\nCONF object needs to be initialized before policy enforcer(). That\nneed to remove cfg.CONF.unregister_opts from TestAuthUtils cleanup\nas this is taken care by cfg.clear() with proper workflow otherwise\nit end up with error\n\"oslo_config.cfg.ArgsAlreadyParsedError: arguments\nalready parsed: reset before unregistering options\"\n\n- https://b132754ee7062a9ab187-9add4719a9922a9385555a8552fc2366.ssl.cf5.rackcdn.com/768520/5/check/openstack-tox-py38/7964354/testr_results.html\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/45102b913db0dbcbe97d24392bcebf45ce5f2697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/45102b913db0dbcbe97d24392bcebf45ce5f2697"}]},"branch":"refs/heads/master"},"1c87ebf9a8a79d990e4a1ba44452b50efae77883":{"kind":"REWORK","_number":8,"created":"2021-03-07 23:50:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/768520/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/768520/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/768520/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/768520/8"}}},"commit":{"parents":[{"commit":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","subject":"Use common rpc pattern for all services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 21:46:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-03-07 23:50:48.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\nCONF object needs to be initialized before policy enforcer(). That\nneed to remove cfg.CONF.unregister_opts from TestAuthUtils cleanup\nas this is taken care by cfg.clear() with proper workflow otherwise\nit end up with error\n\"oslo_config.cfg.ArgsAlreadyParsedError: arguments\nalready parsed: reset before unregistering options\"\n\n- https://b132754ee7062a9ab187-9add4719a9922a9385555a8552fc2366.ssl.cf5.rackcdn.com/768520/5/check/openstack-tox-py38/7964354/testr_results.html\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I1b6c6485bc651fd0b87244a68204036dd4aa37f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/1c87ebf9a8a79d990e4a1ba44452b50efae77883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/1c87ebf9a8a79d990e4a1ba44452b50efae77883"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
