)]}'
{"id":"openstack%2Fmurano~769990","triplet_id":"openstack%2Fmurano~stable%2Ftrain~I4b33c2544c911bd68ed8af4c754c9f5dc0c93be8","project":"openstack/murano","branch":"stable/train","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"I4b33c2544c911bd68ed8af4c754c9f5dc0c93be8","subject":"[stable-only] Cap bandit and fix constraints","status":"MERGED","created":"2021-01-08 20:54:09.000000000","updated":"2021-04-06 19:41:30.000000000","submitted":"2021-04-06 19:40:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769990-cap-bandit-1.6.2","meta_rev_id":"cb6fc23d054d4193ce6df154bf140472f3bbb4d0","_number":769990,"virtual_id_number":769990,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 19:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-04-06 06:36:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-04-06 06:36:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 22:07:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 14:16:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2021-03-22 14:16:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"ed7d4414a61274a0d007a70b3bb02d436e269f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-08 20:54:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c44ae079f4f168060a3a1648b1055e2199d106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 22:07:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/08919e7e136d438eb536639d782a8cb9 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4a17478db5a4e309e2258309a2e33f0 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b91ce80ac134a9b972e96b21fc74b58 : FAILURE in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8db3b0be9814efaad965ddf39e6685a : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c16d03e1604484bbd535b76523f7df6 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e78736eb7ea74ddf988df0aed5a6caaf : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0019da283eed42638b306f52b4d19098 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f0da625e024099b2033402e7ee1eb1 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ab0136f9b24afcaa8e4fe0795019d5 : SUCCESS in 6m 27s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/b1d9289ad3054ba1aa06856429a65763 : FAILURE in 17m 42s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/3584f9d26e964551b936319790c76bc5 : SUCCESS in 54m 48s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/a6f51ca64b1d45558752be0bfa0fc6f8 : FAILURE in 35m 08s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/356fd16289ae4dc4b35d9d503c7410af : SUCCESS in 1h 06m 19s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a2d91fff497f4a74ad0df331a2d55106 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c8cf60bfcedb984cb32235ebce9d2e3bd96a4c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-20 16:13:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fc9735060e6546a366a4b178bca7a9f04a74fae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-20 16:38:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d931ab9816f6dc84307595c450fa9b3ba6e7dc1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 17:30:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b8d32c560084b919c4214fdb7b0dc01 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be1d607388e941e59767186d133ab681 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/093e4a3d4a0541a084ca26948580643e : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa30e4af524a47c399fa3c9c49cdf281 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/769628fa6a24468b8a5bb76a40524989 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a940ace737b44c7a2bcd7db0c24d239 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cecd78b13644aaa9ea1834305e6c1a2 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52547db0d6b54413a1bc31be333839b5 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f97edf6852449b8a372326aa145d50e : SUCCESS in 6m 03s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/4bf106b7348a4b79864e87a28597a311 : FAILURE in 16m 35s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/190d1df5873546e89a7c9328460c22a0 : SUCCESS in 43m 41s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/7d7f44a925e74a8b803ca9a99e7bc57e : FAILURE in 49m 26s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/1c3b6cc76073490a9273571670948b19 : FAILURE in 25m 37s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0a38a37601de41b5bbdfd6f116838d71 : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":3},{"id":"5aea0bfde98749a734fcbd35d847f2fefc8775e4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-22 15:20:15.000000000","message":"Patch Set 3:\n\nStable periodic job of murano is failing [1], this patch fixes stable/train gate.\n\n[1] http://lists.openstack.org/pipermail/openstack-stable-maint/2021-March/thread.html#33494","accounts_in_message":[],"_revision_number":3},{"id":"aa6b95c29eec78536bb94e15111e84aef8b63ecd","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-04-06 06:36:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f440f0b9c6c367be53d211b45b27e5ae8718545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 06:36:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba40947255527ea67858a8d84c58531bb9da2169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 06:55:38.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cbce125a26b4473abca7f6d5b67ae4a5 : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190370b31a1e4cadbae9b48fa2a4b493 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec437f74587404cb1a16515f2a5fe57 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13cab992a1744704b3fef9daebbb1433 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c34c795e0f6f49aea6db1f210a4292fe : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e78d17627a9445b8af65e7a9269c878 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e8c61d128f49978d186e5eb26967ab : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12be004d58304883a804bf4ebf582726 : SUCCESS in 5m 31s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/b803ee1872c64a14a1454d0d5b30f408 : FAILURE in 13m 00s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/79a34ec537a14e0fa4c4e1ada4094a93 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a507d6267b99a23c12f5dc63d1f225d8828633a9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-06 17:56:25.000000000","message":"Patch Set 3:\n\nrecheck - let\u0027s see if the pypi issue is still present","accounts_in_message":[],"_revision_number":3},{"id":"c0189bfa1b7306c096756ffa34fc0ab4efaf7071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:52:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c95fcc75a5c44d9869c583dba665416 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79af5cc760a54dfd8130dbf1c1e2a4a2 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c75cb8e6c45421385bff673a5ea6a15 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3edf18d78254fb38be6479b4a152729 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31ad105fff754498b89b3e25f33c0f1f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb9c43d3b7e4be3bdc6f7c63baf8293 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5f0d515485b4fe39fcb6983ce0b5256 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f808376ee2441a1aa7fd537f540267d : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c28e3022cb7c4d8081ca8508c7bb2363 : SUCCESS in 4m 55s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/985db84e06d34739960c53124e2298bd : FAILURE in 20m 03s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/bfd242a5afdc476ea7a07dea01f62f06 : SUCCESS in 49m 04s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/7ae50be8c8834a39b6cb79478f2494a2 : FAILURE in 38m 33s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/1b3a0b9c74df4dc79a36c4faca856c96 : FAILURE in 37m 45s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d9393159d08d4ed8b71d32e35d737f8f : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":3},{"id":"a5c2c71ce54c81a949e46bb0049ccd0489c6ad89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:52:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1519c880db1c2df75d1996ed648b200960625f7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:40:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/160236236513492a9dbadebf0039ed06 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37198da28aaf427b8e95d102435ec7f4 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab41aa845372409ab3a3ed7d3dbdea28 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b163ca189a884aae8d796c644e21a549 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/343a5383f4a14a23a913a3462526ab4a : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/433add104c864644b9c690bdc3105390 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b91c08804224498a2a177cd0f2f04a5 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617d42540cbb4cfeb32d5156c749426b : SUCCESS in 6m 47s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/5c722de832cc45cb9bd4c45b45dc070b : SUCCESS in 41m 29s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/83aaed3baf1f4460af2dd5e722b72512 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7b9f16f390b3f78d4a26ba1f9ce433e604f81303","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:40:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cb6fc23d054d4193ce6df154bf140472f3bbb4d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:41:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3edbbb1e16de4c9481113f004e53c7c1 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c89d602240c477eb926de51891d6790 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43997de7bda3c0f3dcd4c8ec234167122cadd529","revisions":{"4cedec01cebd97070fad9f56c89c7b0f89a5b0b6":{"kind":"REWORK","_number":1,"created":"2021-01-08 20:54:09.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/769990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/90/769990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/90/769990/1"}}},"commit":{"parents":[{"commit":"4f698865318df4e72b7521aaad34fc0d2ff22dac","subject":"Merge \"Murano api add monkey patch\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4f698865318df4e72b7521aaad34fc0d2ff22dac"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 20:38:06.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 20:51:51.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix constraints","message":"[stable-only] Cap bandit and fix constraints\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nWith the introduction of the new dependency resolver in 20.3 pip the\nlower-constraints job starts to fail as now the constraints are working\nas they should. Multiple fixes needed:\n* lower-constraints job uses the [testenv]install_command, so it gives\n  both the upper-constraints and lower-constraints, which causes the\n  job to fail. As an easy fix install_command is added to this target\n  without the upper constraints.\n* duplicate stestr constraint removed\n* cryptography lower constraint is bumped as with version 2.1 the\n  new resolver of pip takes hours to install the packages\n* contradicting hacking version replaced (in lower-constraints.txt to\n  match with test-requirements.txt), which pulls in newer flake8, too\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I4b33c2544c911bd68ed8af4c754c9f5dc0c93be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4cedec01cebd97070fad9f56c89c7b0f89a5b0b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4cedec01cebd97070fad9f56c89c7b0f89a5b0b6"}]},"branch":"refs/heads/stable/train"},"552b1e05cff54b70bde31a7ca2435a5ad2676056":{"kind":"REWORK","_number":2,"created":"2021-03-20 16:13:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/769990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/90/769990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/90/769990/2"}}},"commit":{"parents":[{"commit":"4f698865318df4e72b7521aaad34fc0d2ff22dac","subject":"Merge \"Murano api add monkey patch\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4f698865318df4e72b7521aaad34fc0d2ff22dac"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 20:38:06.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-03-20 16:12:44.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix constraints","message":"[stable-only] Cap bandit and fix constraints\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nWith the introduction of the new dependency resolver in 20.3 pip the\nlower-constraints job starts to fail as now the constraints are working\nas they should. Multiple fixes needed:\n* lower-constraints job uses the [testenv]install_command, so it gives\n  both the upper-constraints and lower-constraints, which causes the\n  job to fail. As an easy fix install_command is added to this target\n  without the upper constraints.\n* duplicate stestr constraint removed\n* cryptography lower constraint is bumped as with version 2.1 the\n  new resolver of pip takes hours to install the packages\n* contradicting hacking version replaced (in lower-constraints.txt to\n  match with test-requirements.txt), which pulls in newer flake8, too\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I4b33c2544c911bd68ed8af4c754c9f5dc0c93be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/552b1e05cff54b70bde31a7ca2435a5ad2676056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/552b1e05cff54b70bde31a7ca2435a5ad2676056"}]},"branch":"refs/heads/stable/train"},"43997de7bda3c0f3dcd4c8ec234167122cadd529":{"kind":"REWORK","_number":3,"created":"2021-03-20 16:38:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/769990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/90/769990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/769990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/90/769990/3"}}},"commit":{"parents":[{"commit":"4f698865318df4e72b7521aaad34fc0d2ff22dac","subject":"Merge \"Murano api add monkey patch\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4f698865318df4e72b7521aaad34fc0d2ff22dac"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 20:38:06.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-03-20 16:35:51.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix constraints","message":"[stable-only] Cap bandit and fix constraints\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nWith the introduction of the new dependency resolver in 20.3 pip the\nlower-constraints job starts to fail as now the constraints are working\nas they should. Multiple fixes needed:\n* lower-constraints job uses the [testenv]install_command, so it gives\n  both the upper-constraints and lower-constraints, which causes the\n  job to fail. As an easy fix install_command is added to this target\n  without the upper constraints.\n* duplicate stestr constraint removed\n* cryptography lower constraint is bumped as with version 2.1 the\n  new resolver of pip takes hours to install the packages\n* contradicting hacking version replaced (in lower-constraints.txt to\n  match with test-requirements.txt), which pulls in newer flake8, too\n* lower-constraints test job failed with error message:\n  \"[..] unexpected keyword argument \u0027barbican_endpoint_type\u0027\"\n  which is due to old version of castellan.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I4b33c2544c911bd68ed8af4c754c9f5dc0c93be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/43997de7bda3c0f3dcd4c8ec234167122cadd529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/43997de7bda3c0f3dcd4c8ec234167122cadd529"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
