)]}'
{"id":"openstack%2Fmurano~771341","triplet_id":"openstack%2Fmurano~master~Ib42e368cfda2b148a07df0bd74046739f40f7018","project":"openstack/murano","branch":"master","topic":"reuse","hashtags":[],"change_id":"Ib42e368cfda2b148a07df0bd74046739f40f7018","subject":"Use common rpc pattern for all services","status":"MERGED","created":"2021-01-19 05:05:43.000000000","updated":"2021-02-08 01:36:34.000000000","submitted":"2021-02-08 01:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771341-reuse","meta_rev_id":"79cfce6b073f347ad562f382e9ed10eca6b1235d","_number":771341,"virtual_id_number":771341,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 01:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-02-08 00:47:18.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-02-08 00:47:18.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 06:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 00:47:18.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"50eafedf143571a6a55ca2e722c56139f9aee465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 05:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bddf531a92674675b009dab1675b4bef6b8a4ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:01:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/082cbf66990d4f83a24fc981f969613f : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81f3a533282a48dd98ef42f4d457edae : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e460821e644a93aa35bd74ceb3a5de : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a828bd2d24dc47dfa307131e002d3c61 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f8a1c3eb9da403bac7980bbb70df97a : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fdd4bb11d824f49a391f699b5702eb3 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af8786c9f3c47eeb13b3e29495238f8 : SUCCESS in 8m 56s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/8dd03c8d83044c21841bbe4f334d4685 : SUCCESS in 47m 06s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/a6370c8afdf74a428f29ed82a2896b3e : SUCCESS in 49m 59s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/f0cea8705ba345baa52f9cb063febae2 : FAILURE in 41m 32s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/af82a43a6d4240d5b6768e1e96861f21 : SUCCESS in 55m 01s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/681bb2b0f6d1441183d501b96b8a969b : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3039d28539271542dd17574fc99698187a9c7459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 08:03:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5969ee3b9bc7da02b7ad98b4aee50f16481a30d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 08:03:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4e67f4598f5c43e5f257d4e3cfebea28fb10bc69","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 08:08:41.000000000","message":"Topic set to reuse","accounts_in_message":[],"_revision_number":3},{"id":"746fe95ab00a5742bfbd03ee39d1b9a703f9114b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 09:18:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d8af44d4e6f4720a2607e7264d74f6e : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88c075e4a235462a8b09a4ae5d27fd8d : FAILURE in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eada61afe0740e4bbb19d3fdc1c5ffb : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebce6e14f4f34f49848b1c94fb3e1f8d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ff914efddd49b6862b0d7546affede : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22e8ae7be2a0465caaa5fb78cfd4f798 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df420625926a4ba587bd174ad143ab80 : SUCCESS in 8m 19s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/ca10ab7b51a04b888fab2c336b28c7d6 : SUCCESS in 41m 57s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/a3ac8ca106f1463a85d442c3dcfc5dc7 : SUCCESS in 48m 41s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/8786fd855456439988a97fbf50f9e63d : FAILURE in 40m 44s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/6d2860b446f94b75a1e7588497cd44b3 : SUCCESS in 53m 58s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/285ca5ab89154669bbd2b4c7cbf0c53d : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fbab4e9ffa829d1b91b78fa5da090717dae45818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 09:25:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b2616f98d570c65d07f687dc1a3edf101cde2ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:49:27.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe9f5cda56aa4b1d9ff164ad00c04201 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23535c40f207439aa54c5011366a6ead : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ed04826f1346cd87e2cd1fd4e95f90 : FAILURE in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f34175ef9bb4db786c1cba1c3b44ca6 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440dc0fd33a44692bd4b6b2c52da301e : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd13d66089f44f4ea09d5ba4381050d5 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61420af32761431ab16d0acc3ee109f8 : SUCCESS in 9m 39s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/3f8065f99ada42c68bed39a160c27997 : SUCCESS in 50m 11s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/ecb680fb70234b988edd9ba221b88d07 : SUCCESS in 1h 13m 14s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/25b4a15efc0c4911b7de958a9a2fb984 : FAILURE in 58m 26s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/2377bf383a834dee834f4c5674e07416 : SUCCESS in 1h 13m 14s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2e2d532f1ea84db4968571f3b7fc2b26 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":4},{"id":"a81e70238cdcade0d00db63d35b48bf308cda5e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-19 20:24:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b7a71842f72d851f1497f3a92348d3468acd04a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 21:22:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e09fe81278fd48d9ad8a37bc472f3b27 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f66ca2e73c164e85af785f5cdda1148e : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310bad0dca60424584122c81c6af2678 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf2649633d94cddb1d4468d4699c5bf : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb9e6867bab401e88e25a7b00ba0db1 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a036ee0dfe36471b9755fe1703c3fe41 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64751caaa06340f4aefd88a4ab0960cd : SUCCESS in 8m 34s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/aa5a0507ad544e42886f52a8df4bcefc : SUCCESS in 35m 16s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/ee686e7102cf42aca981185bfb853f3c : SUCCESS in 46m 07s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/520c12a76a114b358359c24597d0fe03 : FAILURE in 36m 11s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/b1ffbd6a37c7428e8f37466fdd7e219f : SUCCESS in 48m 35s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/039d0a0f41734d7fbc94fba8817cb9af : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":5},{"id":"b37772e2ae571e3bc94dfb632249ce42c136d138","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-02-08 00:47:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":5},{"id":"6ca2b752d3650d573cbf20335a0d7016e275cfb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 00:47:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"76784621ea9cccc69d57cff74d91f658afd57ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:35:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e799e8c6b4c046e4a1e64a3b89f648ed : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938afe5c24104654beb7814a6003381b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31995a465d544e16941dd57586e1083f : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f64acdc876514891b71e951856d742c1 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b777b3960ba447d2819122c0f841268d : SUCCESS in 7m 43s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/7923d6868dca44a1a25770678c53900e : SUCCESS in 46m 22s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/94b6d40f37c145fca59d3273040f2b22 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c42cf38c5bfc610881a01e2371cdc61febaa46bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:35:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"79cfce6b073f347ad562f382e9ed10eca6b1235d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:36:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3265c43ad2f248c5a0d3cbe49d7960c8 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa","revisions":{"fc46a67200c5bcc668f7219cfdd75bf21b7895eb":{"kind":"REWORK","_number":1,"created":"2021-01-19 05:05:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/771341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/41/771341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/41/771341/1"}}},"commit":{"parents":[{"commit":"e078063cffeab325d98025c43b7a955c53dfc78c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e078063cffeab325d98025c43b7a955c53dfc78c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:04:45.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:05:38.000000000","tz":-480},"subject":"[WIP] Use common rpc pattern for all services","message":"[WIP] Use common rpc pattern for all services\n\nChange-Id: Ib42e368cfda2b148a07df0bd74046739f40f7018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/fc46a67200c5bcc668f7219cfdd75bf21b7895eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/fc46a67200c5bcc668f7219cfdd75bf21b7895eb"}]},"branch":"refs/heads/master"},"661cbf45d8b279cafa54d28039970e897b7e4f28":{"kind":"REWORK","_number":2,"created":"2021-01-19 08:03:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/771341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/41/771341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/41/771341/2"}}},"commit":{"parents":[{"commit":"e078063cffeab325d98025c43b7a955c53dfc78c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e078063cffeab325d98025c43b7a955c53dfc78c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:04:45.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 08:02:04.000000000","tz":-480},"subject":"Use common rpc pattern for all services","message":"Use common rpc pattern for all services\n\nThis patch introduces a common rpc pattern to ensure\nthat the rpc transport is shared where possible. This\nhelps prevent rpc connection leaks and should ensure\nthat we are making the best possible use of all\navailable rpc connections.\n\nChange-Id: Ib42e368cfda2b148a07df0bd74046739f40f7018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/661cbf45d8b279cafa54d28039970e897b7e4f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/661cbf45d8b279cafa54d28039970e897b7e4f28"}]},"branch":"refs/heads/master"},"a7ee4dfd31109715bbd863a4fd8655f9e80ea823":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-19 08:03:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/771341/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/41/771341/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/41/771341/3"}}},"commit":{"parents":[{"commit":"7dd31ad63b4ad5d86d747dbc0e3de18f4a768199","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7dd31ad63b4ad5d86d747dbc0e3de18f4a768199"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:04:45.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 08:03:32.000000000","tz":0},"subject":"Use common rpc pattern for all services","message":"Use common rpc pattern for all services\n\nThis patch introduces a common rpc pattern to ensure\nthat the rpc transport is shared where possible. This\nhelps prevent rpc connection leaks and should ensure\nthat we are making the best possible use of all\navailable rpc connections.\n\nChange-Id: Ib42e368cfda2b148a07df0bd74046739f40f7018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/a7ee4dfd31109715bbd863a4fd8655f9e80ea823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/a7ee4dfd31109715bbd863a4fd8655f9e80ea823"}]},"branch":"refs/heads/master","description":"Rebase"},"7c3fecaad71d0fe32dabe8d4992c16feb543618f":{"kind":"REWORK","_number":4,"created":"2021-01-19 09:25:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/771341/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/41/771341/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/41/771341/4"}}},"commit":{"parents":[{"commit":"7dd31ad63b4ad5d86d747dbc0e3de18f4a768199","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7dd31ad63b4ad5d86d747dbc0e3de18f4a768199"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:04:45.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 09:25:19.000000000","tz":-480},"subject":"Use common rpc pattern for all services","message":"Use common rpc pattern for all services\n\nThis patch introduces a common rpc pattern to ensure\nthat the rpc transport is shared where possible. This\nhelps prevent rpc connection leaks and should ensure\nthat we are making the best possible use of all\navailable rpc connections.\n\nChange-Id: Ib42e368cfda2b148a07df0bd74046739f40f7018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7c3fecaad71d0fe32dabe8d4992c16feb543618f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7c3fecaad71d0fe32dabe8d4992c16feb543618f"}]},"branch":"refs/heads/master"},"e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa":{"kind":"REWORK","_number":5,"created":"2021-01-19 20:24:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/771341/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/41/771341/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/41/771341/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/41/771341/5"}}},"commit":{"parents":[{"commit":"7dd31ad63b4ad5d86d747dbc0e3de18f4a768199","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7dd31ad63b4ad5d86d747dbc0e3de18f4a768199"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 05:04:45.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-01-19 20:24:08.000000000","tz":-480},"subject":"Use common rpc pattern for all services","message":"Use common rpc pattern for all services\n\nThis patch introduces a common rpc pattern to ensure\nthat the rpc transport is shared where possible. This\nhelps prevent rpc connection leaks and should ensure\nthat we are making the best possible use of all\navailable rpc connections.\n\nChange-Id: Ib42e368cfda2b148a07df0bd74046739f40f7018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e5d9d1b74fec1bcf0d93361670a1ffcc22d2bdaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
