)]}'
{"id":"openstack%2Fmurano~794806","triplet_id":"openstack%2Fmurano~stable%2Fwallaby~Ia57b4ce4aa996e61346eae0f4b6a6599f8555652","project":"openstack/murano","branch":"stable/wallaby","topic":"missing-tgtermes.sty","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-10-19 08:48:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia57b4ce4aa996e61346eae0f4b6a6599f8555652","subject":"Add missing font for PDF doc generation","status":"MERGED","created":"2021-06-04 13:13:25.000000000","updated":"2021-10-19 08:50:26.000000000","submitted":"2021-10-19 08:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794806-missing-tgtermes.sty","meta_rev_id":"3dce8e014aef0d48f6ba1bd40f99e75c3710d386","_number":794806,"virtual_id_number":794806,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-19 08:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-10-19 08:50:26.000000000","post_submit":true,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-10-19 07:57:31.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 15:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 11:19:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"4fa5bfe0320892d615f8a2f1cfe72069f37f0a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-04 13:13:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a77c603c1b1aa2a8aac6092694a17fd16e9b11bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:17:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bae5496e21674bafa7ea24e7e7413477 : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeaebb48b44a47b797e158b0f3f4f520 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f55eee1b582436c9b5fb5fb64187706 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39a6f97b61b8416b947d0f05cdc91bea : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/926ac76635ec42e1ab558dc0f40e8daf : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdfb43b09504599896319b4d9129e0e : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32327ae3dab14d6992241b5357b49c76 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32dada5a4325482ca158e985fd3400e6 : SUCCESS in 6m 15s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/de5f77d0c2724555bec1215810faeb60 : SUCCESS in 38m 03s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/8deb286333aa485cbfb2e71518dadc83 : SUCCESS in 47m 48s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/cfd9aa572a044787bae36e58b5d08b77 : FAILURE in 39m 33s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/f929fc913e8543a18b8377b96612baad : SUCCESS in 1h 03m 22s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a16079ecc1e244b7a96e14e34975aa73 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":1},{"id":"dc0eda9ea896c5eb3e1df62865b200491dfdc970","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-14 09:09:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4af83005b339e6e87e0752acf78aa591c35f0fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 09:10:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d085053eadd809720f35e42d21f83e21ea77c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 10:16:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ccd3563f783404ba7552077ce68d22c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f69cdf9d68154b139469bc800d9b1479 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc239f5749a4d839cf1ff38f45b65fa : FAILURE in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eb27cdf8b044bb58fdba8cbbdb17aac : FAILURE in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbd2af67fd94ed79671feb7b5d0aa42 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee264177711041f1a3efe5e344977c9f : SUCCESS in 5m 20s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/01e11d03cd924f6d841b47b283ac8b39 : SUCCESS in 50m 57s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8c2a337cebc746dd91071f8886468ff0 : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":1},{"id":"825662013942b18f6d2a7bfe6a57b6bfa7a3b897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-16 05:55:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c745c0fa6d808c78ee5f6169983a7297ab1b3a44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-16 06:25:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3c9e8710c13ccd0b1fbc4e37b6b80a9fab9ebef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 07:55:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e33bfebb616646278e4315f7fb757f8f : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afa75ba2d72e4b618182a5c0ebb05ac8 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf201424bd3441fbf2117a992452366 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f67320c550694ceca46050215c045dd8 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe35becca844a5484c7f7424ea36cdd : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1205a92b2bdc4863a8a0ebb4e0baa038 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0e1a733e2f402f96c18587588aafbd : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b4b12355c1849179e5f18e46f143615 : SUCCESS in 7m 49s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/3bd30d2e2e9b4a9f8791d3520904f1a6 : SUCCESS in 42m 46s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/2f9fe289a7134a20b908975a356aa023 : SUCCESS in 51m 51s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/5d7f0dda7d734937ade2daeab57217d0 : FAILURE in 40m 35s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/06e4ee2abff04453b9e5ded7f8b00539 : SUCCESS in 57m 26s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/4861ea71314d4730ba5b38c008f1d871 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":3},{"id":"410df77c889f5330067530349b9cacd150f9c28b","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-19 07:57:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6d0257bf6990e39df90f3a026c4e5331dc28d003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 07:58:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f66f72fe5e36d4e6fa5648277d02f61ef7f58bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 08:48:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c405b2ae3d5941579fccb6e06070d3a6 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef87bb1936143e7b1f5df85fefdd450 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b15ba0a38084ebfb5dbb41f77f055a3 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/539bcb698e1c41af8ab7944886184eb0 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f407857c41a46f78b8ae8920d003afa : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/508e0881f01846e696436a856584145e : SUCCESS in 7m 05s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/93c8bb3f54324ea2b4ce2403e64d2dfc : SUCCESS in 42m 45s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5719d0755e824fe096497bfc847a3dfd : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b2f6f9dd220ee0275046677cb3d4c3ca71dc33e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 08:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1acbed10335bd224ff42626a7ab1e83f2d927a1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 08:50:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f5ae0ef2d44e9381bda0f878280983 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6acfc8928f04977a56a1ef0758d6145 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"93dc1f7ce8100d065da5af60820444182f3b8886","revisions":{"0720482ce56da5233c6a429b38dd49c88dc09537":{"kind":"REWORK","_number":1,"created":"2021-06-04 13:13:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/794806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/06/794806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/06/794806/1"}}},"commit":{"parents":[{"commit":"2b3a4f3117d437cd832f010218acf466a8321763","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/2b3a4f3117d437cd832f010218acf466a8321763"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:43:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 13:12:52.000000000","tz":120},"subject":"Add missing font for PDF doc generation","message":"Add missing font for PDF doc generation\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nChange-Id: Ia57b4ce4aa996e61346eae0f4b6a6599f8555652\n(cherry picked from commit d36954704d1e327686596024aeb4c21a370b4d45)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/0720482ce56da5233c6a429b38dd49c88dc09537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/0720482ce56da5233c6a429b38dd49c88dc09537"}]},"branch":"refs/heads/stable/wallaby"},"9d49df5ad945079ab94f0669090c2054d6be67bf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-16 05:55:03.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/06/794806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/06/794806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/06/794806/2"}}},"commit":{"parents":[{"commit":"623d6b49a086fd4fe2d4bdc40720522905ca3a47","subject":"Fix broken unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/623d6b49a086fd4fe2d4bdc40720522905ca3a47"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:43:34.000000000","tz":120},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-10-16 05:55:03.000000000","tz":0},"subject":"Add missing font for PDF doc generation","message":"Add missing font for PDF doc generation\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nChange-Id: Ia57b4ce4aa996e61346eae0f4b6a6599f8555652\n(cherry picked from commit d36954704d1e327686596024aeb4c21a370b4d45)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/9d49df5ad945079ab94f0669090c2054d6be67bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/9d49df5ad945079ab94f0669090c2054d6be67bf"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"93dc1f7ce8100d065da5af60820444182f3b8886":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-16 06:25:50.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/06/794806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/06/794806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/794806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/06/794806/3"}}},"commit":{"parents":[{"commit":"70b8b4899edc9394791768dbaf72bd5d902e047a","subject":"Fix Engine.execute() in func tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/70b8b4899edc9394791768dbaf72bd5d902e047a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:43:34.000000000","tz":120},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-10-16 06:25:50.000000000","tz":0},"subject":"Add missing font for PDF doc generation","message":"Add missing font for PDF doc generation\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nChange-Id: Ia57b4ce4aa996e61346eae0f4b6a6599f8555652\n(cherry picked from commit d36954704d1e327686596024aeb4c21a370b4d45)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/93dc1f7ce8100d065da5af60820444182f3b8886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/93dc1f7ce8100d065da5af60820444182f3b8886"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
