)]}'
{"id":"openstack%2Fmurano~794808","triplet_id":"openstack%2Fmurano~stable%2Fvictoria~Ia57b4ce4aa996e61346eae0f4b6a6599f8555652","project":"openstack/murano","branch":"stable/victoria","topic":"missing-tgtermes.sty","hashtags":[],"change_id":"Ia57b4ce4aa996e61346eae0f4b6a6599f8555652","subject":"Add missing font for PDF doc generation and fix lower constraints","status":"ABANDONED","created":"2021-06-04 13:18:10.000000000","updated":"2024-04-11 17:32:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"082a084cb991ac6346af6e0829209dd986f03bba","_number":794808,"virtual_id_number":794808,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2021-08-14 09:55:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-08-14 09:10:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-08-14 09:10:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 15:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 11:20:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"1bf45b982b1c79a5c3b5a603e91767fd54251df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-04 13:18:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"850bc0c6e26abd69f432bb4263a079a8dc38f029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:10:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ebd0bc25f644332a7ee1b8f104ab87f : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54fd4d0faf224f0cb6e8f7b5c5c29a1c : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec0b7647ddc45e6b7725ca6916d33ef : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f32b0b7d984d4fc99813eb6cd993c00d : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e24325ac95e44df9bddbbc64ffae848 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa78cae1bc1b4598a1b9b24ae9db24da : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe6b59c2e3f64003bfccc22fea8930a8 : SUCCESS in 5m 42s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/39d29b1e2f564fe9adfd870914dc4104 : SUCCESS in 36m 24s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/d061b5dd8f184911ad1e17d2888c4439 : SUCCESS in 49m 58s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/a6f94266900a46658a39e6f9cf95f49e : FAILURE in 47m 25s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/cd722869b6c7479a9660bae8226e643b : SUCCESS in 55m 10s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1888665267408fa88e77366c76e3c6 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"de5ecfbbb2eafc6125141037f9f906b170afe646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-04 19:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bedfb668e660589c81bf009edd0928894a713d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:09:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/200e19503c654255b19274ff51aa49a7 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca45985a6be541ca87dde7caeed86295 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e80af5b29fe4aed844b7ce5713a8212 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a2e209f625471196fa48e1023765dc : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d75b35c1ab640ffaa35ca9f81c46e77 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e36f916cf8414f14aa6ac2c661662751 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f7289fbef5469ab425d9359bab8a04 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f7f709985b42e6b5ef60e9db92972f : SUCCESS in 5m 42s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/74ad2ec42c1247ac97fb17e7e4ccebad : SUCCESS in 37m 59s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/cfcdb6e1a7f9470d804ad6df72dded9a : SUCCESS in 47m 17s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/882e451e01074149afa7b462e8e8d425 : FAILURE in 38m 15s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/8924dea279564d6ca94f2ecbc94bc613 : SUCCESS in 56m 45s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f2f3ad7c10a342c6a2f45c6d3840bb18 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":2},{"id":"19a0cea68a5bba71747befd30b54a1ab9c9cb90a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-14 09:10:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8e31547c3d0d3b0990ed604e923e2bb9d018ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 09:10:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eab52f8f9425d29d80cdd8531e002f080fe6d231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 09:55:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/348a4044068745e290c7ef6d7cafdb53 : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/104f4d7896f94f6a9d828915b6acaba6 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d7062cc74a4254bc1805d5af772920 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c07e294b50246d8bbdea96957543766 : FAILURE in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f032d7a8d8346858467799a1be736c6 : FAILURE in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77576142a33a4ed69130e356c13f8398 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79acbf8161eb4859a9c35f3a335216e8 : SUCCESS in 5m 51s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/a505037abdf840e4b0aab4b153aaa0e8 : FAILURE in 31m 33s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8cfb6354f5df41a2b2ad419936d9385f : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":2},{"id":"082a084cb991ac6346af6e0829209dd986f03bba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-11 17:32:29.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/murano transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a09bb038c0dbb4935b88c1cc9b30ea5cfd5d2d3","revisions":{"8ba61f7a5d3baf6ffa693128fcd6b6ffd164d39e":{"kind":"REWORK","_number":1,"created":"2021-06-04 13:18:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/08/794808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/08/794808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/08/794808/1"}}},"commit":{"parents":[{"commit":"b8f0708149dac18b72e30ed47860f3f3993d54c5","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b8f0708149dac18b72e30ed47860f3f3993d54c5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:43:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 13:13:42.000000000","tz":120},"subject":"Add missing font for PDF doc generation","message":"Add missing font for PDF doc generation\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nChange-Id: Ia57b4ce4aa996e61346eae0f4b6a6599f8555652\n(cherry picked from commit d36954704d1e327686596024aeb4c21a370b4d45)\n(cherry picked from commit 0720482ce56da5233c6a429b38dd49c88dc09537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ba61f7a5d3baf6ffa693128fcd6b6ffd164d39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ba61f7a5d3baf6ffa693128fcd6b6ffd164d39e"}]},"branch":"refs/heads/stable/victoria"},"7a09bb038c0dbb4935b88c1cc9b30ea5cfd5d2d3":{"kind":"REWORK","_number":2,"created":"2021-06-04 19:07:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/08/794808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/08/794808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/08/794808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/08/794808/2"}}},"commit":{"parents":[{"commit":"b8f0708149dac18b72e30ed47860f3f3993d54c5","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b8f0708149dac18b72e30ed47860f3f3993d54c5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:43:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 19:04:53.000000000","tz":120},"subject":"Add missing font for PDF doc generation and fix lower constraints","message":"Add missing font for PDF doc generation and fix lower constraints\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nWith the new pip resolver (introduced in pip 20.3) the constraints\nchecking is now working better, thus reveals multiple constrain\nconflicts. This patch also fixes those problems.\n\nChange-Id: Ia57b4ce4aa996e61346eae0f4b6a6599f8555652\n(cherry picked from commit d36954704d1e327686596024aeb4c21a370b4d45)\n(cherry picked from commit 0720482ce56da5233c6a429b38dd49c88dc09537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7a09bb038c0dbb4935b88c1cc9b30ea5cfd5d2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7a09bb038c0dbb4935b88c1cc9b30ea5cfd5d2d3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
