)]}'
{"id":"openstack%2Fmurano~805514","triplet_id":"openstack%2Fmurano~stable%2Fvictoria~I5268132018e0a283bd35b5599cf8ca41968dde93","project":"openstack/murano","branch":"stable/victoria","topic":"bug/1938676-stable/victoria","attention_set":{},"removed_from_attention_set":{"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2022-01-06 11:15:13.000000000","reason":"zhurong replied on the change"}},"hashtags":[],"change_id":"I5268132018e0a283bd35b5599cf8ca41968dde93","subject":"Fix broken unit test","status":"MERGED","created":"2021-08-22 03:40:46.000000000","updated":"2022-01-06 12:18:59.000000000","submitted":"2022-01-06 12:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805514-bug/1938676-stable/victoria","cherry_pick_of_change":803151,"cherry_pick_of_patch_set":1,"meta_rev_id":"8ce76021e998df0b2426eb9666fc9af2318f6707","_number":805514,"virtual_id_number":805514,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 12:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2022-01-06 11:15:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2022-01-06 11:15:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 03:40:46.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-08-22 03:54:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea3efeffe63f33b9c50d32ce13bdfe4c268d273a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-22 03:40:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"69d1041d6de52f46195a996d780c1f2c3da163a3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-22 03:41:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ac6c41128046093daf4bc229909ab8d3c12df2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 03:54:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/645cd8143fe54dcd81d73c83d500fbf8 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c213a1616774bcea357052df6a7d362 : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac3e6ba69094265bb0811e888a850ca : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a41f885dbc545e98ad60b01c236b555 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91da593b4f594f48bbaaeab4efbb216d : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ebc5372ae64a289af971d880c14588 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fe1e87473cb47a09eb8e74d893f3fc4b3dd5e4c1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-18 01:52:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed4ad3b511f8ca448ce3b7912d89f713a02d77a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 02:07:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1074a1b8548b45f6975af925962e6df8 : FAILURE in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3117c5cfcb9140a8a5b56224f8cab2ca : FAILURE in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f0b3cee77a492294e1300c8c473596 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb34bb66f734a9a8fb95c66c9abc53a : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7318634e3de465c93d5e000cb9fcc6e : FAILURE in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ea455f241247dc91b0581c880e7409 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"30aca0c54de05b0817f4e6d0303f97e383bf65af","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-01-06 09:24:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e56dd60b4a6ea5c06d36ff6da19d8c6fd8e4c4b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-01-06 09:37:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8bd693b6161df04a5d2ea0fb99e6f4f4eaadc091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 09:37:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a216a4c1d2804ce29272525206bdae28 : FAILURE in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cfa5a6969ac42f79e322cabe48202d5 : FAILURE in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c24c17d28d7476f9b207bc969ae96ab : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3bd64a73fd3438eb5a3cc3cfb73f2d1 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed619a7323bc4c68bbec70c486bfbd54 : FAILURE in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfbaf3d7b83462881b55cb662eb199b : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2a967f002ff5898cb03fe97a69567006bbc4f38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-01-06 09:38:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"378256d476084d18fdc763f267eaa35fa2ce3914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 10:42:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bf82e5fe8fc4fd7b4570d922014aaf8 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ac8cf813944dffabfa1cc902aa4854 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d182cebace445eac2f0017d3175189 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f18873e450542ff91a25d91f7291295 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222173d8e2974f6f958f2b6af00d858b : SUCCESS in 9m 13s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/b7f47875aa8c4b34b9a368824550a140 : SUCCESS in 42m 37s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/3e2b2b20670847aa9445d746f8c1609e : SUCCESS in 49m 40s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/984ca9dbc6db4f6d8daf47526d70a57c : FAILURE in 38m 48s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/5708aa67cb1247558edbfb87d7c7dc65 : SUCCESS in 57m 19s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/768be7ca478d4cf5976d6a64d63fa16e : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":3},{"id":"576d6798a6b80f80c38fe3c8c0b8f9f49e5891ac","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-01-06 11:15:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea26f054bae6335752098574b3671789fee3f28b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 11:16:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1421ad716eb792ed58adb78c7cf055c5e1641fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:13:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8939ba4a1aa741ce95fa0ef988896183 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9636119a013420194a4386985eae31d : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5141fce076b46aa97914cbbd48ae2f5 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8d57aac7fd4d208af66a1061f4f62a : SUCCESS in 9m 17s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/304f960cfb554264beb309902960eff2 : SUCCESS in 51m 20s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/465121029c674c839142b52be9133b53 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f10741e962a7d6ba4f18d037d9de3578adc8b401","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:13:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8ce76021e998df0b2426eb9666fc9af2318f6707","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:18:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26097f78be0f47fdae1eb4a61c049f8c : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ec727dfc6bbfba3a9f1260110b0297c557e1295","revisions":{"2ae131ccdee9c6e8ddd1c2785cba8ce16b3d1ebf":{"kind":"REWORK","_number":1,"created":"2021-08-22 03:40:46.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/14/805514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/14/805514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/14/805514/1"}}},"commit":{"parents":[{"commit":"b8f0708149dac18b72e30ed47860f3f3993d54c5","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b8f0708149dac18b72e30ed47860f3f3993d54c5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-02 08:20:27.000000000","tz":540},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-08-22 03:40:46.000000000","tz":0},"subject":"Fix broken unit test","message":"Fix broken unit test\n\nThis change mocks sqlalchemy.update during unit tests properly to avoid\nthe following error.\n\nsqlalchemy.exc.ArgumentError: subject table for an INSERT, UPDATE or\nDELETE expected, got \u003cMagicMock name\u003d\u0027models.Instance\u0027 id\u003d\u0027...\u0027\u003e\n\nCloses-Bug: #1938676\nChange-Id: I5268132018e0a283bd35b5599cf8ca41968dde93\n(cherry picked from commit 2800f0b62c2936cd68c35623010c3f57d58b975d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/2ae131ccdee9c6e8ddd1c2785cba8ce16b3d1ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/2ae131ccdee9c6e8ddd1c2785cba8ce16b3d1ebf"}]},"branch":"refs/heads/stable/victoria"},"06c75649beba88aad7d5106b688c36d28fed6449":{"kind":"REWORK","_number":2,"created":"2022-01-06 09:37:25.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/14/805514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/14/805514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/14/805514/2"}}},"commit":{"parents":[{"commit":"b8f0708149dac18b72e30ed47860f3f3993d54c5","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b8f0708149dac18b72e30ed47860f3f3993d54c5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-02 08:20:27.000000000","tz":540},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2022-01-06 09:37:24.000000000","tz":0},"subject":"Fix broken unit test","message":"Fix broken unit test\n\nThis change mocks sqlalchemy.update during unit tests properly to avoid\nthe following error.\n\nsqlalchemy.exc.ArgumentError: subject table for an INSERT, UPDATE or\nDELETE expected, got \u003cMagicMock name\u003d\u0027models.Instance\u0027 id\u003d\u0027...\u0027\u003e\n\nCloses-Bug: #1938676\nChange-Id: I5268132018e0a283bd35b5599cf8ca41968dde93\n(cherry picked from commit 2800f0b62c2936cd68c35623010c3f57d58b975d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/06c75649beba88aad7d5106b688c36d28fed6449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/06c75649beba88aad7d5106b688c36d28fed6449"}]},"branch":"refs/heads/stable/victoria"},"8ec727dfc6bbfba3a9f1260110b0297c557e1295":{"kind":"REWORK","_number":3,"created":"2022-01-06 09:38:44.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/14/805514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/14/805514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/14/805514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/14/805514/3"}}},"commit":{"parents":[{"commit":"b8f0708149dac18b72e30ed47860f3f3993d54c5","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/b8f0708149dac18b72e30ed47860f3f3993d54c5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-02 08:20:27.000000000","tz":540},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2022-01-06 09:38:44.000000000","tz":0},"subject":"Fix broken unit test","message":"Fix broken unit test\n\nThis change mocks sqlalchemy.update during unit tests properly to avoid\nthe following error.\n\nsqlalchemy.exc.ArgumentError: subject table for an INSERT, UPDATE or\nDELETE expected, got \u003cMagicMock name\u003d\u0027models.Instance\u0027 id\u003d\u0027...\u0027\u003e\n\nCloses-Bug: #1938676\nChange-Id: I5268132018e0a283bd35b5599cf8ca41968dde93\n(cherry picked from commit 2800f0b62c2936cd68c35623010c3f57d58b975d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ec727dfc6bbfba3a9f1260110b0297c557e1295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ec727dfc6bbfba3a9f1260110b0297c557e1295"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
