)]}'
{"id":"openstack%2Fmurano~809720","triplet_id":"openstack%2Fmurano~stable%2Fxena~I94a6faa2bc7e79048d4b23e53889c8cbdc80fc76","project":"openstack/murano","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I94a6faa2bc7e79048d4b23e53889c8cbdc80fc76","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-09-17 14:16:48.000000000","updated":"2021-10-18 02:35:16.000000000","submitted":"2021-10-18 02:33:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809720-create-xena","meta_rev_id":"b0a61c45b9d59c31edc56193b0e2438b68b45134","_number":809720,"virtual_id_number":809720,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 02:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2021-10-18 01:31:54.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2021-10-18 01:31:54.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 14:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 01:31:54.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"5c693ce078e486007103ee7662c96fb38a7eb786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-17 14:16:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f96f6fd777d45f7b8bbfae814bb00e4e87fdf0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:45:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a8008e02504446961ba9b411a46389 : FAILURE in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7833e69f3c6e4af9a713b31609766480 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e988af3d814a1abe9dbc6a0b1200c8 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb308807334452096e3ca3a2600ec25 : FAILURE in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0579e2a9e74adabb616818e4a7d0c5 : FAILURE in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8c07e95eb64ef5abf7a34e5ce9dd75 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2863f17e664f15bff537b6bdbf33da : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b94583afca1fb6792ef498a31dd589150ab0cb58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-16 05:54:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cdb9be5e58748e259f154215e0e7d5fc8113564a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 07:15:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/687d9f2416bc4fcca83a1c349447e1ab : SUCCESS in 14m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05846e4c5af647a48ba48f225f3f2b8c : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0532f7446b4e4eabbbce5708b29fbf9f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc3ebf223d347ed8b96fa1cd9d2289a : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33d9a13e1e9e4b0f90eaaecd116c7041 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dd3c0102889428a847a586f49f17679 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8988a295544b4bcba673b756b01fd39c : SUCCESS in 8m 58s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/91e9f2814d744f64b896a837804ec7a1 : SUCCESS in 49m 24s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/15ff4739d0024330b14fed38744a7efe : SUCCESS in 56m 36s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/015279f4e6434eb398cc8d44f68ae509 : FAILURE in 46m 39s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/c1ae030e3f4d4c7dbe168a096049f40f : SUCCESS in 1h 15m 27s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5229410301fa4b889f83f8fa76a4fde0 : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3eaafaebc8046709662284a470dee1746e720fc1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-10-18 01:31:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c30de3a6493769ca7489aeaaecda6be7b388b391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 01:32:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a1f919c10367cd60f67b2f7784c4d642baccff7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 02:33:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749fbf54c64d407c985efdfcdd02c2a5 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20530a55bbbf446994d585686d67f542 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd1ca0bed09428ea4be1ac2880c1287 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eb7a786e73c435c900c0d09c50343c9 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293f165b0b014bd696586a8e28060e54 : SUCCESS in 9m 03s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/29e04ebabddf41d4a9ce1e03805a78dd : SUCCESS in 48m 06s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/965864542e5d4cc192eadbe78e0e0f9b : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":2},{"id":"07250f48f2bfcca4f8e55910f236cd461cc1c016","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 02:33:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b0a61c45b9d59c31edc56193b0e2438b68b45134","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 02:35:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f423db9702f4487b69fdfc0eefeb31d : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b63349ca05bcb279a79fca6d3878b8f7deade61","revisions":{"934692a120d6500b4509954e24879156e221b1f6":{"kind":"REWORK","_number":1,"created":"2021-09-17 14:16:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/20/809720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/809720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/809720/1"}}},"commit":{"parents":[{"commit":"8ab430b2f4963e985f3f0de066bb8ddf1e89b847","subject":"Merge \"Enable tls-proxy support for test jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8ab430b2f4963e985f3f0de066bb8ddf1e89b847"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 14:16:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 14:16:34.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I94a6faa2bc7e79048d4b23e53889c8cbdc80fc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/934692a120d6500b4509954e24879156e221b1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/934692a120d6500b4509954e24879156e221b1f6"}]},"branch":"refs/heads/stable/xena"},"7b63349ca05bcb279a79fca6d3878b8f7deade61":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-16 05:54:13.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/20/809720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/20/809720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/20/809720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/20/809720/2"}}},"commit":{"parents":[{"commit":"15fd7fbd3a2e3166f8255edf0b256d2c06a3f48a","subject":"Fix Engine.execute() in func tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/15fd7fbd3a2e3166f8255edf0b256d2c06a3f48a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 14:16:34.000000000","tz":0},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-10-16 05:54:13.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I94a6faa2bc7e79048d4b23e53889c8cbdc80fc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7b63349ca05bcb279a79fca6d3878b8f7deade61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7b63349ca05bcb279a79fca6d3878b8f7deade61"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
