)]}'
{"id":"openstack%2Fmurano~832569","triplet_id":"openstack%2Fmurano~stable%2Fyoga~Ie04fa9f3f79e0256bc98af077458b055a46ec1e3","project":"openstack/murano","branch":"stable/yoga","topic":"create-yoga","hashtags":[],"change_id":"Ie04fa9f3f79e0256bc98af077458b055a46ec1e3","subject":"Update .gitreview for stable/yoga","status":"MERGED","created":"2022-03-08 12:05:41.000000000","updated":"2022-03-30 01:05:07.000000000","submitted":"2022-03-30 01:03:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832569-create-yoga","meta_rev_id":"af9866eacda1df48014f3cb7ae0eeaaedf662524","_number":832569,"virtual_id_number":832569,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 01:03:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2022-03-29 23:24:43.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2022-03-29 23:24:43.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 13:51:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-03-08 14:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-29 23:24:43.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"af2a105bb90d9cedb2b6549e95c99713d5242152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-08 12:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07eaf5122667224fa23cdd79b6ba3ba3158fa99a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-08 13:51:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b903a1569d8444cb13fd5c02409783080abef9bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:12:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84929078263b449bb5493f6acb0e3805 : FAILURE in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3291c2650eb04d84a9bce34a27b887c7 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2241b1d9848b48b0b42f0393035c95ab : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6adee87337b84f0493f65150c6480db2 : FAILURE in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/020b14df2cf44ddb9343ccbf388f4b05 : FAILURE in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7436cc949243149230b73ca522c8f0 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cb029d6ca9f2c810ce8338cd6686a0706f713720","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-03-29 23:24:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62f71864993f758c7b0aade2ad6db4ac5a2b782a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 00:14:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d53d54f17e4cf1badd8ad2827b76e0 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fd762575ff46e9b892a0fa4686e314 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae957ebc441417283717d667a1fd9c1 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30bb12825593420fbf70abcfc70951a3 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43dd6ed9197c4cfaaffbb0d896e247fb : SUCCESS in 11m 08s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/e6c31daeefcb461d931dffd6894537d4 : SUCCESS in 29m 35s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/ec2ad44a4b084d15abef08b229961a6a : SUCCESS in 47m 26s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/c873ede5218c4315aebb59d8928e934a : FAILURE in 34m 16s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/cecddbe66e8d4d54852198f11acc2be8 : SUCCESS in 46m 44s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/53d97e88abf442908e7bd2e6ca4550da : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d037a6531089775749a216383f947dccd48d51a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 00:14:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4aedd93209bc5e2d47a59e5206e7b78fb16f4cdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 01:03:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e74e6a9605547238bd4d3b800428cee : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97f32440278a4766bb61f8aa5a362aed : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc6a2e5d72cc4cc0aa8cf02c10e24597 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f9590ddc7641759ab07d175b8fc7a3 : SUCCESS in 10m 47s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/41b2da0e494a43ab93e0c226af6140ca : SUCCESS in 48m 18s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8512bbb5cb334ce8939f1c7cd44a4ab3 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3b22ee781c1fb6af829db1338c2269bd94db2820","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 01:03:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af9866eacda1df48014f3cb7ae0eeaaedf662524","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 01:05:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730c4169f84948448c302ed84cc64ec5 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"719070f087d7c41f351fd45557ba4491d8681f59","revisions":{"719070f087d7c41f351fd45557ba4491d8681f59":{"kind":"REWORK","_number":1,"created":"2022-03-08 12:05:41.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/832569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/69/832569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/69/832569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/69/832569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/69/832569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/69/832569/1"}}},"commit":{"parents":[{"commit":"f0258fc8438fd4bd3c48d7c8ba4a4e1166a219bf","subject":"Merge \"Remove the unused paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f0258fc8438fd4bd3c48d7c8ba4a4e1166a219bf"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 12:05:40.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 12:05:40.000000000","tz":0},"subject":"Update .gitreview for stable/yoga","message":"Update .gitreview for stable/yoga\n\nChange-Id: Ie04fa9f3f79e0256bc98af077458b055a46ec1e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/719070f087d7c41f351fd45557ba4491d8681f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/719070f087d7c41f351fd45557ba4491d8681f59"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
