)]}'
{"id":"openstack%2Fmurano~835506","triplet_id":"openstack%2Fmurano~stable%2Fyoga~I2830ccf840bf4d1d8a516287adee51c46d2a5583","project":"openstack/murano","branch":"stable/yoga","topic":"context.project_id-patch1-stable/yoga","attention_set":{},"removed_from_attention_set":{"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2022-03-29 16:13:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2830ccf840bf4d1d8a516287adee51c46d2a5583","subject":"Tenant to project migration for RequestContext","status":"MERGED","created":"2022-03-29 13:36:45.000000000","updated":"2022-03-29 16:34:13.000000000","submitted":"2022-03-29 16:13:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"835506-context.project_id-patch1-stable/yoga","cherry_pick_of_change":835488,"cherry_pick_of_patch_set":7,"meta_rev_id":"ac8fbbf76e1153e8f69893602a98faffd8311258","_number":835506,"virtual_id_number":835506,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-29 16:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-03-29 13:57:42.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-03-29 13:51:32.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-03-29 13:51:32.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 13:36:45.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-03-29 13:57:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-03-29 15:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2482fa8016eda66f2d514d4fcb922e6bcac656d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-03-29 13:36:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bc733e7ac4edc7259175edea88fa8ccd6b0b709a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-03-29 13:51:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9cdb16008513fceb7e86c3be45a01faac174b4b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-29 13:57:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4075d0f1f4eee6fc67839a10fac8b907df8716b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c4f59950e046d98f2452df1ed51a37 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f60a6894f844b8e88a2872206f88bd6 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adfc098c17a949839559180036569cd0 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6d00bf4f64f4cbe865038967c56a3eb : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41e5b4fe3e84943a80008aba9f11751 : SUCCESS in 10m 01s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/4823f519e2f5400ab3bc8b61366d85e6 : SUCCESS in 42m 14s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/13e5a5deeebe4ee9814a66a81564f033 : SUCCESS in 51m 58s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/5bfaa63bcef74af9846707110c06c154 : FAILURE in 46m 26s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/703f8a013d2249da881c29e5d0c0f202 : SUCCESS in 49m 47s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c5e49c82f00a4b47ae4878a6cf8b4afb : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"67d1b4111fe3aa22c882080f0e0d88d4cf5ac53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:09:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec3c2bbc35ce9ed2b50e93aebdccefdbeed54db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:13:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338f797a7b354a47b7d21282325f0e79 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5396d97bcc64ec88270b593fadfaa13 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c12433831f94fa69d207ccfc404d349 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77194fa488d4b578ffe648b7a5b7c3c : SUCCESS in 12m 34s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/195480079067456080a6f229a743185f : SUCCESS in 51m 54s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/509a652c5c164ee3bfdded872c9e30f1 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b53128677023d506a6cd6780ff48d4397768ca1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:13:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac8fbbf76e1153e8f69893602a98faffd8311258","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:34:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e65c0723b1c4e5f8cb1653a9c52e547 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f994a7d4e939713c16432374b1509bd178e70a0","revisions":{"8f994a7d4e939713c16432374b1509bd178e70a0":{"kind":"REWORK","_number":1,"created":"2022-03-29 13:36:45.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/06/835506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/06/835506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/835506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/835506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/06/835506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/06/835506/1"}}},"commit":{"parents":[{"commit":"f0258fc8438fd4bd3c48d7c8ba4a4e1166a219bf","subject":"Merge \"Remove the unused paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f0258fc8438fd4bd3c48d7c8ba4a4e1166a219bf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-03-28 19:26:18.000000000","tz":0},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2022-03-29 13:36:45.000000000","tz":0},"subject":"Tenant to project migration for RequestContext","message":"Tenant to project migration for RequestContext\n\nAs of the oslo.context 4.0.0 release, RequestContext objects no\nlonger have a tenant attribute, instead expecting callers to use\nproject_id for that purpose. Update all context.tenant references to\ncontext.project_id in order to facilitate this transition.\n\nChange-Id: I2830ccf840bf4d1d8a516287adee51c46d2a5583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8f994a7d4e939713c16432374b1509bd178e70a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/8f994a7d4e939713c16432374b1509bd178e70a0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
