)]}'
{"id":"openstack%2Fmurano~839890","triplet_id":"openstack%2Fmurano~stable%2Fyoga~I1d80bfc5f9c3fecc8551ffc8592e39f76b671485","project":"openstack/murano","branch":"stable/yoga","topic":"fix-docs-use-uc","hashtags":[],"change_id":"I1d80bfc5f9c3fecc8551ffc8592e39f76b671485","subject":"[CI] Install dependencies for docs target","status":"ABANDONED","created":"2022-04-29 09:57:56.000000000","updated":"2024-02-26 09:41:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63bb7456e51f8812e5db84ee4a38555c1a8f1d6f","_number":839890,"virtual_id_number":839890,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-29 11:00:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 11:00:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b681df1abcad0eb8afc59989c1ce0290a8c3e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-29 09:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4ab65a2f1127fcd21f7cf626ea6e6920cf86781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-29 09:58:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3db0c0c59062da6a123226893a2c44d0e3edb47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:00:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963abb5f71904c9d919954eee5bb6553 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d78107fd56234f31b303d2d4b2d15072 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec99cfc07df4d308babc892e1e3279e : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebdf0f0ab36b415c8e70e2ef17cc9117 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1677d953b6f400dba75fb141844e36e : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ad046fbd334d06bef0c98fcdd5b409 : SUCCESS in 4m 21s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/df5560118c7b49e599f659f50650092f : SUCCESS in 35m 27s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/2d80fac2b54d4314b1640d7663bcf449 : SUCCESS in 38m 10s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/5adbf56ad55f4446a14c1c9eb1355eb3 : FAILURE in 34m 40s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/adf18be72d664e4ca0f1f5ec264e4407 : SUCCESS in 48m 21s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/bdc52f60104943d08b6175f26e09e906 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":2},{"id":"63bb7456e51f8812e5db84ee4a38555c1a8f1d6f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 09:41:17.000000000","message":"Abandoned\n\nthis branch is moving to unmaintained","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d447b05e0ccf538bb646d8a3e8b523f1d1f70d2","revisions":{"44ebadb86947e285625c4472327a2f856a297586":{"kind":"REWORK","_number":1,"created":"2022-04-29 09:57:56.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/839890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/90/839890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/90/839890/1"}}},"commit":{"parents":[{"commit":"27dd423015152ca07f22d4bb396e2d5e4a0f940b","subject":"Merge \"Update .gitreview for stable/yoga\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/27dd423015152ca07f22d4bb396e2d5e4a0f940b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 17:51:40.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-29 09:57:09.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs murano\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nChanges:\n  tox.ini\n\nNOTE(elod.illes): branch specific upper constraints file link updated.\n\nChange-Id: I1d80bfc5f9c3fecc8551ffc8592e39f76b671485\n(cherry picked from commit 1f211823aa8766075d80694c1c3eb6f8c4a3dedf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/44ebadb86947e285625c4472327a2f856a297586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/44ebadb86947e285625c4472327a2f856a297586"}]},"branch":"refs/heads/stable/yoga"},"3d447b05e0ccf538bb646d8a3e8b523f1d1f70d2":{"kind":"REWORK","_number":2,"created":"2022-04-29 09:58:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/839890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/90/839890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/90/839890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/90/839890/2"}}},"commit":{"parents":[{"commit":"27dd423015152ca07f22d4bb396e2d5e4a0f940b","subject":"Merge \"Update .gitreview for stable/yoga\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/27dd423015152ca07f22d4bb396e2d5e4a0f940b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 17:51:40.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-29 09:58:43.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs murano\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nChanges:\n  tox.ini\n\nNOTE(elod.illes): branch specific upper constraints file link updated.\n\nChange-Id: I1d80bfc5f9c3fecc8551ffc8592e39f76b671485\n(cherry picked from commit 1f211823aa8766075d80694c1c3eb6f8c4a3dedf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/3d447b05e0ccf538bb646d8a3e8b523f1d1f70d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/3d447b05e0ccf538bb646d8a3e8b523f1d1f70d2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
