)]}'
{"id":"openstack%2Fmurano~839892","triplet_id":"openstack%2Fmurano~stable%2Fwallaby~I1d80bfc5f9c3fecc8551ffc8592e39f76b671485","project":"openstack/murano","branch":"stable/wallaby","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-04-11 17:37:01.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1d80bfc5f9c3fecc8551ffc8592e39f76b671485","subject":"[CI] Install dependencies for docs target","status":"ABANDONED","created":"2022-04-29 10:01:07.000000000","updated":"2024-04-11 17:37:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"91ff3a6a8f8ef7cae05f89e5cbc2e9af3f15b082","_number":839892,"virtual_id_number":839892,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-29 11:14:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 11:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc9d088f768ef5ea87b1f0303ea8ad19f6d81ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-29 10:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29996179f8f102ee418c99b5d5b01703bb2e5ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:14:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4689eb16cf5f4662812569407bceb31a : FAILURE in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dd359b5c49045679a6a4d5d271f3e9c : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e8e471442743e79ed1344e94e2923b : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/283255e4fd6f41b9a2c1e660107d5c40 : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dafa4effcd84d43ba8b7c1155722079 : FAILURE in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0938f1f8a14c423d8ac6e15ca2e3bff2 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3d59877a6d4f1ab23ec548b3b0e127 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb2df1948f0e486db447054cbc74063a : SUCCESS in 5m 40s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/4db9f17f8a514946beee28795d22026b : SUCCESS in 35m 53s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/cd6e2b1d973a436d90026f41bfd4264f : SUCCESS in 45m 52s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/e80b47340f7e454fb2482e59022b8895 : FAILURE in 37m 28s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/41000c9ee76745a8bb50c6a8608ae292 : SUCCESS in 1h 05m 31s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/00678d432f404525a280faabaa4367e6 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"91ff3a6a8f8ef7cae05f89e5cbc2e9af3f15b082","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-11 17:37:01.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/murano transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0510f022958e232ae857f256861c506af38c1e41","revisions":{"0510f022958e232ae857f256861c506af38c1e41":{"kind":"REWORK","_number":1,"created":"2022-04-29 10:01:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/92/839892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/92/839892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/839892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/839892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/92/839892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/92/839892/1"}}},"commit":{"parents":[{"commit":"93dc1f7ce8100d065da5af60820444182f3b8886","subject":"Add missing font for PDF doc generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/93dc1f7ce8100d065da5af60820444182f3b8886"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 17:51:40.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-29 10:00:49.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs murano\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nChanges:\n  tox.ini\n\nNOTE(elod.illes): branch specific upper constraints file link updated.\n\nChange-Id: I1d80bfc5f9c3fecc8551ffc8592e39f76b671485\n(cherry picked from commit 1f211823aa8766075d80694c1c3eb6f8c4a3dedf)\n(cherry picked from commit 3d447b05e0ccf538bb646d8a3e8b523f1d1f70d2)\n(cherry picked from commit 433ca35839dd6db7eeffa158bec4cf69ec48f415)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/0510f022958e232ae857f256861c506af38c1e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/0510f022958e232ae857f256861c506af38c1e41"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
