)]}'
{"id":"openstack%2Fmurano~851455","triplet_id":"openstack%2Fmurano~stable%2Fyoga~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/yoga","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:06:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:10:51.000000000","updated":"2022-08-12 15:11:30.000000000","submitted":"2022-08-12 15:06:38.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851455-unmaintained-stable-periodic","meta_rev_id":"b9267925a2ef1849f6cf17ba3e1a9227e1039e71","_number":851455,"virtual_id_number":851455,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:38.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 20:12:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:38.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:06:38.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 20:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:06:38.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"dc661e8e9147b8fae3cdb779419a01618e390f7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1673d5046d7ebdf34ba26315fe57e2d1cf49893e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:11:24.000000000","message":"Topic set to unmaintained-stable-periodic","accounts_in_message":[],"_revision_number":1},{"id":"b39286810c59139b7f4eccc873bb7ac87879648d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 20:12:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01d12dd89522469fa779130b61ffda99\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48148403851b489d83275a23a90860d5 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21050707ea2445db94e3d00a79eeffe5 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c755311725fb4f41b4ea50fb25855711 : FAILURE in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3a6429091094160b845dc08047ca7ac : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2768a431f34d6597c749e71e3c8d83 : SUCCESS in 10m 54s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/0c75a521c80a4f5592311fbca8512d13 : SUCCESS in 46m 48s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/8b6c57889b9b49d0b75a541d5f9b56c7 : SUCCESS in 50m 37s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/62ce498677344a8aaa90887c3592cb68 : FAILURE in 51m 41s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/8831d2b3f6e2454db46edee07d16cba8 : SUCCESS in 49m 27s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8bdb50ad5e3e48edb5c054ed079a9fc0 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a37f2fbede5b95af49849e344e1fc4b6837c5da5","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:38.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"2b01317fcd074e0510589b26786e9f72268f6c7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b9267925a2ef1849f6cf17ba3e1a9227e1039e71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:11:30.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e1c11869f6c461a979e9c1bdb7afbe0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe5eca3a34541a898f7dad895adc6c2 : FAILURE in 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b3211357b60e7200e5efe48e5cfc89bf6e8f278","revisions":{"6b3211357b60e7200e5efe48e5cfc89bf6e8f278":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:10:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/851455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/55/851455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/55/851455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/55/851455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/55/851455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/55/851455/1"}}},"commit":{"parents":[{"commit":"27dd423015152ca07f22d4bb396e2d5e4a0f940b","subject":"Merge \"Update .gitreview for stable/yoga\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/27dd423015152ca07f22d4bb396e2d5e4a0f940b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/6b3211357b60e7200e5efe48e5cfc89bf6e8f278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/6b3211357b60e7200e5efe48e5cfc89bf6e8f278"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
