)]}'
{"id":"openstack%2Fmurano~851456","triplet_id":"openstack%2Fmurano~stable%2Fxena~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/xena","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:07:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:14:15.000000000","updated":"2022-08-12 15:11:31.000000000","submitted":"2022-08-12 15:07:08.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851456-unmaintained-stable-periodic","meta_rev_id":"c7ebbecfeab469d41842dbb5032cfed6eca2de48","_number":851456,"virtual_id_number":851456,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:07.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 20:18:56.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:07.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:07:07.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 20:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:07:07.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"b3ce36774219adf4dee3880ed49dda6bfe63301f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfa76b81763eb7ae40af1d5836a2c2faa8f033e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 20:18:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c726ad009c9f420c963bf74875b4b674\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f19f056176ee4e1aa0ee4481ff96da04 : FAILURE in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e58249b4fc9243ecb79cee2f054fef59 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5df068a6774480b9b6a08a6015b1df9 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1caee905c2146a3b93d69bf6bc097c6 : FAILURE in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1a74d5017943eabc94c5e1562e4448 : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5574955cf40d448ca96f395a57cd5c46 : FAILURE in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83011fc1ef0f4f7484b04fc06c519346 : SUCCESS in 8m 39s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/162806290a8943d8bd0a0482d4d71cff : SUCCESS in 52m 42s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/db4c8d3324184e4b9ac06a7fbd6bcb36 : SUCCESS in 53m 11s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/84394424c14940f196abadf45afcb1ec : FAILURE in 46m 22s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/002e6274cc5640ef82bce773ad23bd41 : SUCCESS in 1h 03m 45s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/39b8597f6b7f4ca4b1d841903b64f788 : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3b547e3ee465dc2cff7d7b7aa7feec9d55e0f8e1","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:07.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"75264621cc8368afad632078de2302f102065920","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7ebbecfeab469d41842dbb5032cfed6eca2de48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:11:31.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7081270c37a49ecb1f3f807a9d7bdbf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1a4d0b634f4c0d926f4599a0ee6f51 : FAILURE in 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20874f920e484643e430c91a3b6f7036bc7d7820","revisions":{"20874f920e484643e430c91a3b6f7036bc7d7820":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:14:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/851456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/56/851456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/56/851456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/56/851456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/56/851456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/56/851456/1"}}},"commit":{"parents":[{"commit":"7b63349ca05bcb279a79fca6d3878b8f7deade61","subject":"Update .gitreview for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/7b63349ca05bcb279a79fca6d3878b8f7deade61"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:13:22.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n(cherry picked from commit 6b3211357b60e7200e5efe48e5cfc89bf6e8f278)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/20874f920e484643e430c91a3b6f7036bc7d7820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/20874f920e484643e430c91a3b6f7036bc7d7820"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
