)]}'
{"id":"openstack%2Fmurano~851459","triplet_id":"openstack%2Fmurano~stable%2Fussuri~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/ussuri","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:06:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:18:00.000000000","updated":"2022-08-12 15:11:31.000000000","submitted":"2022-08-12 15:06:48.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851459-unmaintained-stable-periodic","meta_rev_id":"f1b996899a7ba00b086d1b1ad93aa4610b115003","_number":851459,"virtual_id_number":851459,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:48.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 20:13:12.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:48.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:06:48.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 20:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:06:48.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"a490de728ead915ad4e77ed32fec0ead15db4f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:18:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8879a4635bb3cee421d3c44106ad9361db6332ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 20:13:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7de5e473b3b24033b750ad7296e86ecc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cefdf20b98444ce5b94b9fc76da899a8 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/581c9d9bf7b843838e461c5e03cf3c7a : RETRY_LIMIT in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706b5bd6187f41f5ae448b41333a04a2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dff40c201ff4ddd814ec3b4ccc2d665 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ee4b003725a437f8ed9731029367afd : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169b2b6922a649ef8bd8c9e21673f96a : SUCCESS in 8m 46s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/2ded91662bf54f92b643e7c08d0533e8 : SUCCESS in 36m 29s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/9f76818a20b24e9ba169c796c3667008 : SUCCESS in 45m 37s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/bf21c60c05fe4586bbe3326b0d18b12e : FAILURE in 34m 39s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/11995d0d5f7941008e2627abfc35fbe6 : SUCCESS in 53m 09s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/750750f7618746f793546e3d38b06479 : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":1},{"id":"593924a5e472af9e8011ba08fb36a4cf206ff3d2","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:48.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"29653e0ff609bc7dda2ea1fc726dd7f84075e212","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1b996899a7ba00b086d1b1ad93aa4610b115003","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:11:31.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e63074434024114aea5abfa442f0ec6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03888d6422374eaeb844f65148ec93a8 : FAILURE in 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44f52b51150b556f27ee0141d6a8e05c25adc199","revisions":{"44f52b51150b556f27ee0141d6a8e05c25adc199":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:18:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/59/851459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/59/851459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/59/851459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/59/851459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/59/851459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/59/851459/1"}}},"commit":{"parents":[{"commit":"1daccfced80ae3dfac59f08fd20d31244ef9d94a","subject":"Add native grenade zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/1daccfced80ae3dfac59f08fd20d31244ef9d94a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:17:36.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n(cherry picked from commit 6b3211357b60e7200e5efe48e5cfc89bf6e8f278)\n(cherry picked from commit 20874f920e484643e430c91a3b6f7036bc7d7820)\n(cherry picked from commit 972686aa3eb527021b832e56a5d2f7c43f43f15a)\n(cherry picked from commit 0bd8f3dcbb39d8e7cbb747ee77aa97bd5c9ad515)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/44f52b51150b556f27ee0141d6a8e05c25adc199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/44f52b51150b556f27ee0141d6a8e05c25adc199"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
