)]}'
{"id":"openstack%2Fmurano~851460","triplet_id":"openstack%2Fmurano~stable%2Ftrain~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/train","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:07:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:19:15.000000000","updated":"2022-08-12 15:12:08.000000000","submitted":"2022-08-12 15:07:46.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851460-unmaintained-stable-periodic","meta_rev_id":"86b46fc9f69b47f261553c50d4079747be2390fc","_number":851460,"virtual_id_number":851460,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:46.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 22:18:01.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:46.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:07:46.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 22:18:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:07:46.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"f9dfb71f842c1ed2a990bcaee005e54a40ff56e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c0ee6532bdb818fb80154f226d5110b5b5c983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 22:18:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39b2f9661d734fa3bcf7de9fcdb015cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8454d19f134834ba6351341ef79b22 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6c6d0e51f1a4390ab5bfdfebe3e96bb : RETRY_LIMIT in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3f60a6181d41efb7cf140476e27d5e : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8863b175f48149ecb01fa8c096849c6d : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/103459bcd10f477cb663a956232c43ef : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65f43783dd37480083dd4dac570430b4 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222fdb91b3d8451cb9de1b07bfd2b206 : SUCCESS in 9m 51s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/98a9072b38234ea4921b2400fa26a71b : FAILURE in 19m 40s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/f5ddcda8895a486689bbdaafa70e9dd3 : SUCCESS in 46m 24s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/72aacdc3653c464b93b13b55c12ba17a : FAILURE in 38m 35s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/0871be12ee5c4401a3974fdd86a5cb68 : FAILURE in 2h 56m 23s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2cc49c79da8f4b55bdfa96dedf5bc6d0 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":1},{"id":"18d71286b6600e9afba8599f2c5e173e0b37946f","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:46.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"cebadd52c6ee2ea8609e44cfe746966dd65b926f","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86b46fc9f69b47f261553c50d4079747be2390fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:12:08.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af08346f6d544289bb49e03b5ca1a68a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba7759314834daca8beab6bd27e7e56 : FAILURE in 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92dcce2d9d08c221a8ae81c639e207113de7a136","revisions":{"92dcce2d9d08c221a8ae81c639e207113de7a136":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:19:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/60/851460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/60/851460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/60/851460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/60/851460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/60/851460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/60/851460/1"}}},"commit":{"parents":[{"commit":"43997de7bda3c0f3dcd4c8ec234167122cadd529","subject":"[stable-only] Cap bandit and fix constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/43997de7bda3c0f3dcd4c8ec234167122cadd529"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:18:53.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n(cherry picked from commit 6b3211357b60e7200e5efe48e5cfc89bf6e8f278)\n(cherry picked from commit 20874f920e484643e430c91a3b6f7036bc7d7820)\n(cherry picked from commit 972686aa3eb527021b832e56a5d2f7c43f43f15a)\n(cherry picked from commit 0bd8f3dcbb39d8e7cbb747ee77aa97bd5c9ad515)\n(cherry picked from commit 44f52b51150b556f27ee0141d6a8e05c25adc199)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/92dcce2d9d08c221a8ae81c639e207113de7a136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/92dcce2d9d08c221a8ae81c639e207113de7a136"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
