)]}'
{"id":"openstack%2Fmurano~851463","triplet_id":"openstack%2Fmurano~stable%2Frocky~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/rocky","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:06:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:23:30.000000000","updated":"2022-08-12 15:10:29.000000000","submitted":"2022-08-12 15:06:27.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851463-unmaintained-stable-periodic","meta_rev_id":"04bd5b6b6550b84480f612dc5eaab66ba6a47fb0","_number":851463,"virtual_id_number":851463,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 20:09:04.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:06:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:06:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 20:09:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:06:27.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"62444c7c2be286da995b85e4bd5436e750a6a9e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:23:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d657236882a227853ab01b729711e4b076be8db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 20:09:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34e68d4e9d69450c9eccaf7afea121af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b660c4a3745746a3a314a1d7e7d997c2 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/739b890e0b0241139f88499d3b36548a : FAILURE in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebb06c79860340f3bd42483e8e3a2622 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa631392b87e483083f047a000bd8d3f : SUCCESS in 6m 20s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/5c3b7f3ae28a4a87a4966f8689019bb8 : FAILURE in 20m 07s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/9df484194b4a463bbc7cae2f69e1087d : FAILURE in 40m 39s (non-voting)\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/7b74f073ddb441869c47aad48c1ea984 : FAILURE in 32m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d843f71ed1c34665bb9a87f4514eb612 : RETRY_LIMIT in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b22ce471f1604892b22e181a038b24db : SUCCESS in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f66fffbabb0f17a0dc8f0ff1b2a16e6245c4383","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:27.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"beec17acd077ea717ce692eeeea7ae43be4dec81","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:06:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04bd5b6b6550b84480f612dc5eaab66ba6a47fb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:10:29.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/694cf707469942728971714ff810f688\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43a9c45d90946d08ca58001e46dfa3a : FAILURE in 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d310e44e06fe6cfc19d3c4f4bf93803fe69abf5b","revisions":{"d310e44e06fe6cfc19d3c4f4bf93803fe69abf5b":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:23:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/851463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/63/851463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/63/851463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/63/851463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/63/851463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/63/851463/1"}}},"commit":{"parents":[{"commit":"82dc0eb1e1fd88e491abfaab71fd9812edf736fb","subject":"Fix murano ci UT error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/82dc0eb1e1fd88e491abfaab71fd9812edf736fb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:22:56.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n(cherry picked from commit 6b3211357b60e7200e5efe48e5cfc89bf6e8f278)\n(cherry picked from commit 20874f920e484643e430c91a3b6f7036bc7d7820)\n(cherry picked from commit 972686aa3eb527021b832e56a5d2f7c43f43f15a)\n(cherry picked from commit 0bd8f3dcbb39d8e7cbb747ee77aa97bd5c9ad515)\n(cherry picked from commit 44f52b51150b556f27ee0141d6a8e05c25adc199)\n(cherry picked from commit 92dcce2d9d08c221a8ae81c639e207113de7a136)\n(cherry picked from commit 587ccfb23113c46110070ef6478962e61b6130db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d310e44e06fe6cfc19d3c4f4bf93803fe69abf5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/d310e44e06fe6cfc19d3c4f4bf93803fe69abf5b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
