)]}'
{"id":"openstack%2Fmurano~851464","triplet_id":"openstack%2Fmurano~stable%2Fqueens~I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","project":"openstack/murano","branch":"stable/queens","topic":"unmaintained-stable-periodic","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-12 15:07:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4","subject":"[stable-only] Remove periodic-stable-jobs template","status":"MERGED","created":"2022-07-28 19:24:05.000000000","updated":"2022-08-12 15:11:32.000000000","submitted":"2022-08-12 15:07:27.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851464-unmaintained-stable-periodic","meta_rev_id":"06efad4b70c2382c218057992f003c72d8df584f","_number":851464,"virtual_id_number":851464,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 21:37:06.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2022-08-12 15:07:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2022-08-12 15:07:27.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 21:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 15:07:27.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"f8ba8b780749a8772d87a3ab1933530a424af2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 19:24:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4590bc07019798c4e49e544ec531e90a1e3e294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 21:37:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c535f8d4beaf4fa0a1b2bfc62baabb09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e75380219341c3b0035b7ac69d7454 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/760ff770d7c340f1b3a27d5434b28385 : FAILURE in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9527469318c34bab80d8b8c32e5dd079 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a51af3f6d5ae4aeea9f722322a858f21 : SUCCESS in 5m 16s\n- murano-rally https://zuul.opendev.org/t/openstack/build/8448ec4dff354210bb9ff6b072b3d22a : FAILURE in 2h 07m 09s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/f16517a577f54a0f96f0a9322f0eb457 : FAILURE in 2h 05m 38s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/e530bd9667244c9d92a3c6f09009f911 : FAILURE in 2h 08m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a42d708865f74a88b206d80a7f8d3926 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"652cb40db96b561f568738438d7bbfaceb15e305","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:27.000000000","message":"Patch Set 1: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"b75f18d5696555575dd045cf1f7306ec93b5e05a","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2022-08-12 15:07:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"06efad4b70c2382c218057992f003c72d8df584f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:11:32.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/390880fe6a2d4e7aa715e81514659921\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1f81932ff44aa985c37679baccb022 : FAILURE in 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4677afec3ab0fd10e52d65057f8bb3c40b8a8a4","revisions":{"e4677afec3ab0fd10e52d65057f8bb3c40b8a8a4":{"kind":"REWORK","_number":1,"created":"2022-07-28 19:24:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/64/851464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/64/851464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/64/851464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/64/851464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/64/851464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/64/851464/1"}}},"commit":{"parents":[{"commit":"e36e91f53ad2f156debae3a2fc8acaad023f4638","subject":"Fix murano ci UT error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e36e91f53ad2f156debae3a2fc8acaad023f4638"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:10:04.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-07-28 19:23:45.000000000","tz":120},"subject":"[stable-only] Remove periodic-stable-jobs template","message":"[stable-only] Remove periodic-stable-jobs template\n\nPeriodic stable jobs are failing for a long time since the project is\nnot actively maintained. This patch removes the periodic-stable-jobs\ntemplate to free up CI resource.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I56428c7fed44c8e0970e4ae5c4a46a3c96bdcca4\n(cherry picked from commit 6b3211357b60e7200e5efe48e5cfc89bf6e8f278)\n(cherry picked from commit 20874f920e484643e430c91a3b6f7036bc7d7820)\n(cherry picked from commit 972686aa3eb527021b832e56a5d2f7c43f43f15a)\n(cherry picked from commit 0bd8f3dcbb39d8e7cbb747ee77aa97bd5c9ad515)\n(cherry picked from commit 44f52b51150b556f27ee0141d6a8e05c25adc199)\n(cherry picked from commit 92dcce2d9d08c221a8ae81c639e207113de7a136)\n(cherry picked from commit 587ccfb23113c46110070ef6478962e61b6130db)\n(cherry picked from commit d310e44e06fe6cfc19d3c4f4bf93803fe69abf5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e4677afec3ab0fd10e52d65057f8bb3c40b8a8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/e4677afec3ab0fd10e52d65057f8bb3c40b8a8a4"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[]}
