)]}'
{"id":"openstack%2Fmurano~858638","triplet_id":"openstack%2Fmurano~master~I963e741078514b1cca6ed0e510cefaeccdd3499a","project":"openstack/murano","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2022-10-12 03:55:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I963e741078514b1cca6ed0e510cefaeccdd3499a","subject":"Fix compatibility with oslo.db 12.1.0","status":"MERGED","created":"2022-09-21 07:36:23.000000000","updated":"2022-10-12 03:57:57.000000000","submitted":"2022-10-12 03:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858638","meta_rev_id":"633823beabe44190dece9e7483d3c1abd87d23a7","_number":858638,"virtual_id_number":858638,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 03:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"all":[{"value":2,"date":"2022-10-12 03:00:01.000000000","_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"all":[{"value":1,"date":"2022-10-12 03:00:01.000000000","_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 00:23:29.000000000","updated_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"reviewer":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"state":"REVIEWER"},{"updated":"2022-10-11 02:48:16.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"},{"updated":"2022-10-12 03:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08b2083988e0319f600270c021fe0c6dad38dbeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-09-21 07:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9c863494ae888293ec772a0552952ec819d4303","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-11 00:23:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64ca9ebfcddb729c93cd2c704e77fffe6e6ea844","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-11 02:48:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7de214026dc0b5bed72e9180faf0370aa19b91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-12 02:12:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14026e87c63c6504fa10a079d341131752500a76","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-12 03:00:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01e74e4bad6ce5ded865f5368726c82ded0e7459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:13:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f01df93b5d34b7eb3f5308b019c3565\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/00fa9d3e5e894fb5a08d85ac5777bdcc : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/620cdb65c008494ca6d287306e900f06 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7716a5bf57433da4afc6788707d1ce : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb186b066f74d73bcba9e91f997d10e : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59b5a3af761c461b8cea771fbc146154 : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f94d07e8b4374dc499b9bb7032fbfd2e : FAILURE in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c42820230304e8c8680d215e4f04488 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d08783cc1bd94bfbad2aed37500f0b6c : SUCCESS in 5m 12s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/20b0557a1cb24f40b929f411b5f2a03e : SUCCESS in 36m 41s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/3f52af4fc21a4745bcfe3160905cdfa4 : SUCCESS in 48m 26s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/0debcd4d449a4aeb883ba5c02328ebc1 : FAILURE in 37m 32s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/263688d112bb4250b44078b7f926f18e : FAILURE in 35m 59s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/677f96a3d2e9425f8049a5099f0ae8ec : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e68d22cd74133e7adec438f36756c73ab3b11ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:13:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23851ec7b4a91800c4a868437348f36cc371fb08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:55:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b81aba1f06de406a811cd6f372852e2c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d2bad32576f4134bdd722d487dfaceb : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6582e087ccde4bf2a21778458a441cc1 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb9d05b087947f1b2129c1128b52056 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/247416f285ee4ebdb31d6bd28abebffe : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0782f7195340a889f6cd9dcdf52493 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab781008e7847b09e411871e58acd66 : SUCCESS in 4m 29s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/94b73fa04a3e4fe7bef2d241fc867459 : SUCCESS in 41m 16s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d0c77aea61b3427099664db433334403 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b096ba9da278ac727edcc7d063670b1afc8b44c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:55:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"633823beabe44190dece9e7483d3c1abd87d23a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 03:57:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34e611ba777a46c69baa1362abd409c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651d400c51c5490981beb93276bce490 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46b34c05b0d382bbdb5f635a9589d49dc8b839ea","revisions":{"52d20b27c33e5e669c696a4c31be4992187797e9":{"kind":"REWORK","_number":1,"created":"2022-09-21 07:36:23.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/38/858638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/38/858638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/38/858638/1"}}},"commit":{"parents":[{"commit":"4434b57486eda599a15624dd27eca7ef18bcfa1f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4434b57486eda599a15624dd27eca7ef18bcfa1f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-09-21 07:26:07.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-09-21 07:36:20.000000000","tz":120},"subject":"Fix compatibility with oslo.db 12.1.0","message":"Fix compatibility with oslo.db 12.1.0\n\noslo.db 12.1.0 has changed the default value for the \u0027autocommit\u0027\nparameter of \u0027LegacyEngineFacade\u0027 from \u0027True\u0027 to \u0027False\u0027. This is a\nnecessary step to ensure compatibility with SQLAlchemy 2.0. However, we\nare currently relying on the autocommit behavior and need changes to\nexplicitly manage sessions. Until that happens, we need to override the\ndefault.\n\nNote that this patch was written using this one as example:\nhttps://review.opendev.org/c/openstack/magnum/+/858374\n\nChange-Id: I963e741078514b1cca6ed0e510cefaeccdd3499a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/52d20b27c33e5e669c696a4c31be4992187797e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/52d20b27c33e5e669c696a4c31be4992187797e9"}]},"branch":"refs/heads/master"},"46b34c05b0d382bbdb5f635a9589d49dc8b839ea":{"kind":"REWORK","_number":2,"created":"2022-10-12 02:12:49.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/38/858638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/38/858638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/38/858638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/38/858638/2"}}},"commit":{"parents":[{"commit":"4434b57486eda599a15624dd27eca7ef18bcfa1f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4434b57486eda599a15624dd27eca7ef18bcfa1f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-09-21 07:26:07.000000000","tz":120},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2022-10-12 02:11:52.000000000","tz":660},"subject":"Fix compatibility with oslo.db 12.1.0","message":"Fix compatibility with oslo.db 12.1.0\n\noslo.db 12.1.0 has changed the default value for the \u0027autocommit\u0027\nparameter of \u0027LegacyEngineFacade\u0027 from \u0027True\u0027 to \u0027False\u0027. This is a\nnecessary step to ensure compatibility with SQLAlchemy 2.0. However, we\nare currently relying on the autocommit behavior and need changes to\nexplicitly manage sessions. Until that happens, we need to override the\ndefault.\n\nNote that this patch was written using this one as example:\nhttps://review.opendev.org/c/openstack/magnum/+/858374\n\nAlso include Zuul gate fix. Fix as described in [1][2]\n\n[1] https://lists.opendev.org/pipermail/service-announce/2022-September/000044.html\n[2] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I963e741078514b1cca6ed0e510cefaeccdd3499a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/46b34c05b0d382bbdb5f635a9589d49dc8b839ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/46b34c05b0d382bbdb5f635a9589d49dc8b839ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"}}]}],"submit_requirements":[]}
