)]}'
{"id":"openstack%2Fmurano~861074","triplet_id":"openstack%2Fmurano~stable%2Fzed~I963e741078514b1cca6ed0e510cefaeccdd3499a","project":"openstack/murano","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"18418":{"account":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"last_update":"2022-10-18 03:56:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I963e741078514b1cca6ed0e510cefaeccdd3499a","subject":"Fix compatibility with oslo.db 12.1.0","status":"MERGED","created":"2022-10-12 10:29:22.000000000","updated":"2022-10-18 07:33:27.000000000","submitted":"2022-10-18 03:56:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861074","meta_rev_id":"61e4069a8cc9bffd325e435034633b9bc3e7e30d","_number":861074,"virtual_id_number":861074,"owner":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 03:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"recommended":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"all":[{"value":1,"date":"2022-10-13 21:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":2,"date":"2022-10-18 02:55:24.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-18 07:33:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":1,"date":"2022-10-18 02:55:24.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 10:29:22.000000000","updated_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"CC"},{"updated":"2022-10-12 11:34:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 02:55:24.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2022-10-18 07:33:27.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"68f49d7130c37832d3602dacbc0cba4904a45bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-12 10:29:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b064e3196cbb7d6ae7a23a536434e64c0ebb3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:34:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/317913b525554b179e839bf01858d6f0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb7e28ed114a4b0dae5e6c9c5b53db5e : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c154e5704a634583b9919e1f88f1988e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79bed39f5554221a071beb45bcbdc6f : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35e947f802224a17895e6f86cd519327 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cff576ec4c24028be421650bf2aeb0c : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/612bef6785fd48cc8f5c0df1c8b6591a : FAILURE in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880c904e88354b8b981f6b87f31fd5b8 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dbe6133a6354c5998e01e7b45a850be : SUCCESS in 5m 30s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/6f9a6c5807d34de1bb465190faef0e11 : SUCCESS in 25m 57s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/5844b506178c4764be605cb25652b47a : SUCCESS in 50m 24s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/3ea25621d852417da0120632d2293679 : FAILURE in 29m 24s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/fa99d3b222cc427f867af893e20dcb30 : SUCCESS in 59m 05s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6b51e05b092f45b6a9feb48373b03227 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c80c39c7deab9830a8aa2e184e8f7be9be272f90","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2022-10-13 21:54:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"274ded3ba4ea17d505bdcdb9abfa4f267b3077d3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-10-18 02:55:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f52f0ab5dab232546017f60f1568086fd69eb3ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 02:56:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73b28bf4cf79d93c9873d82ac3e64ff358b89d10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:56:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/589d758f428f47cc96d8ebfd2d2482ef\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7ec1283f64c4497a6bca9051ed1798e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d99edef215849169ec916ac19d2687a : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9032f2fe79854303a44da74a828ac458 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/314a26b23ae44166a34499a1a7372158 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afff91023174adaab9d3bd1327d57ab : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cf361f0757f4b59a8d8b106e866fbd0 : SUCCESS in 5m 37s\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/19c9d34fd522432096fb04d0e3beb802 : SUCCESS in 43m 14s\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1b233c9e3ec94627a29d0999a5361813 : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1dab8aa7795cc674a378cc34d75403b807470997","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:56:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2eead96dfe110f2709753191b66ed4400080fc3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:58:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c1360516735483c8e2f01b14d1cfaec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ce5c51fb184fc5aa39010e5d57fb4f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"61e4069a8cc9bffd325e435034633b9bc3e7e30d","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-10-18 07:33:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"153bb0ea837741638765c80ddd08ed9ba5dfd560","revisions":{"153bb0ea837741638765c80ddd08ed9ba5dfd560":{"kind":"REWORK","_number":1,"created":"2022-10-12 10:29:22.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/74/861074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/74/861074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/74/861074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/74/861074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/74/861074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/74/861074/1"}}},"commit":{"parents":[{"commit":"4434b57486eda599a15624dd27eca7ef18bcfa1f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/4434b57486eda599a15624dd27eca7ef18bcfa1f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-09-21 07:26:07.000000000","tz":120},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2022-10-12 10:28:00.000000000","tz":660},"subject":"Fix compatibility with oslo.db 12.1.0","message":"Fix compatibility with oslo.db 12.1.0\n\noslo.db 12.1.0 has changed the default value for the \u0027autocommit\u0027\nparameter of \u0027LegacyEngineFacade\u0027 from \u0027True\u0027 to \u0027False\u0027. This is a\nnecessary step to ensure compatibility with SQLAlchemy 2.0. However, we\nare currently relying on the autocommit behavior and need changes to\nexplicitly manage sessions. Until that happens, we need to override the\ndefault.\n\nNote that this patch was written using this one as example:\nhttps://review.opendev.org/c/openstack/magnum/+/858374\n\nAlso include Zuul gate fix. Fix as described in [1][2]\n\n[1] https://lists.opendev.org/pipermail/service-announce/2022-September/000044.html\n[2] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I963e741078514b1cca6ed0e510cefaeccdd3499a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/153bb0ea837741638765c80ddd08ed9ba5dfd560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/153bb0ea837741638765c80ddd08ed9ba5dfd560"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[]}
