)]}'
{"id":"openstack%2Fmurano~904673","triplet_id":"openstack%2Fmurano~master~Ie53dc343927bc107a2445ed96a84d24d6513dc7d","project":"openstack/murano","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-15 03:49:39.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-05-15 03:49:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie53dc343927bc107a2445ed96a84d24d6513dc7d","subject":"Fix python 3.11 job and update python classifier in setup.cfg","status":"ABANDONED","created":"2024-01-04 07:58:32.000000000","updated":"2024-05-15 03:49:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"57e154105e2c17414a5af199f9a0f0ed50c79fa5","_number":904673,"virtual_id_number":904673,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-22 21:10:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 08:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 03:10:39.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"d2aa3557205bfa3e3e91d9b57f054725dd3f7f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-04 07:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e971c220a7c88f03f029a37179daaf47df4a769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 08:22:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/471d68d49b0d4ad8be59af3bc91c45c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff853f5880b344cd9092e6be99a043af : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5862e270f044992bb436a56bb743462 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d514e8cddf77468fb372016250d651e8 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67c9ba673b1a484a9b183f3b220a163e : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5ac3797b8564e0abac96c989084a20d : RETRY_LIMIT in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9ae0be4d1c4dd89cace275bd9d328e : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5f1fbc63946a7f755ee4c5986e0425b35f37461f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-05 03:10:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5cff5836e83ac6b9729702198ec2563ad4513cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-05 06:01:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9ad24158fd10ae0ad37d314acf581fab1f124138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 08:08:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccb22a00f6f34f5183c6c6c5321d53ba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44ef8379eb72405495b25c5439d10c7a : SUCCESS in 4m 42s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5321fb6e8a9a4b83a74c3d6b4ce87d49 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f011c25812b4eeaa1bd14494581ab61 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e789c37a4c0d444899e803cd59c2bbe0 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1143b39ec31425f8bd2c99b271d77b7 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0594f0990e0e47d9b1a359e3712f8648 : FAILURE in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978f3b32c0604cba8838b8a9b5cf0d73 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8caa9af20f5d49c8be6edc39a1a1e321 : SUCCESS in 4m 47s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/f8dc054550984965afe67c53dbf63f84 : FAILURE in 2h 01m 15s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/ef1027705b284cf2b012d53772bcefb2 : FAILURE in 48m 49s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/f83f3baefc70430083b4e1cf258caf82 : FAILURE in 34m 25s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/d4cabeb41b8845789a2e1b8065c2f86d : FAILURE in 45m 16s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/213bfd85f7a548fba661de5a4bc39c29 : FAILURE in 34m 04s","accounts_in_message":[],"_revision_number":2},{"id":"221b3173a3beb5caa9057a568d8bbf251b4ed439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-22 20:14:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7cb507de3472ab149571c103cf1c11adfa9b9618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 21:10:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22498d6f47af4080b556572edb7f30e2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ecc1ba23bc1b4c4dbb5b03b3cb3446ca : SUCCESS in 3m 31s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3a779c70ad148a6b79c5f9a9428ef38 : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2366219ae0154a779d308d65a3d701c9 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0bfcb06f334a7cac08d9b4fdb3d634 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fecfd23a42664b479b9d243ffa8e7daa : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ba731837aab43cca7cf8eed48afb5d1 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17ed71e34c64d4bb01afc1dcae9f861 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c45dc751fc64df6b143054513b1d03f : SUCCESS in 5m 24s\n- murano-rally-task https://zuul.opendev.org/t/openstack/build/5b948b2c0b3e4b118b61d505a50ef631 : SUCCESS in 39m 12s (non-voting)\n- murano-tempest-api https://zuul.opendev.org/t/openstack/build/bc3604d7be834390becf928510348ff3 : FAILURE in 50m 13s\n- murano-tempest-cfapi https://zuul.opendev.org/t/openstack/build/1f265d1ab5c64aa2aec5410d016ace2b : FAILURE in 32m 38s (non-voting)\n- murano-grenade https://zuul.opendev.org/t/openstack/build/fd79259bb0084957a477ff2e6a2bf278 : FAILURE in 40m 03s (non-voting)\n- murano-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8bf9f407bdf04847a03c6e67c850bc1b : FAILURE in 49m 00s","accounts_in_message":[],"_revision_number":3},{"id":"57e154105e2c17414a5af199f9a0f0ed50c79fa5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 03:49:39.000000000","message":"Abandoned\n\nMurano project has been retired https://review.opendev.org/c/openstack/governance/+/919358","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"26668baf7b9fa62dd0f0df57d6e1d1226510e39d","revisions":{"fa73f617f22782ab4a79634987a467cfd7f9e1d4":{"kind":"REWORK","_number":1,"created":"2024-01-04 07:58:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/904673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/73/904673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/73/904673/1"}}},"commit":{"parents":[{"commit":"c898a310afbc27f12190446ef75d8b0bd12115eb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/c898a310afbc27f12190446ef75d8b0bd12115eb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:13:31.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:58:30.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nChange-Id: Ie53dc343927bc107a2445ed96a84d24d6513dc7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/fa73f617f22782ab4a79634987a467cfd7f9e1d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/fa73f617f22782ab4a79634987a467cfd7f9e1d4"}]},"branch":"refs/heads/master"},"f749fad38861b41503361b70f8ca030e5c56ff16":{"kind":"REWORK","_number":2,"created":"2024-01-05 06:01:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/904673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/73/904673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/73/904673/2"}}},"commit":{"parents":[{"commit":"c898a310afbc27f12190446ef75d8b0bd12115eb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/c898a310afbc27f12190446ef75d8b0bd12115eb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:13:31.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-05 06:00:41.000000000","tz":-480},"subject":"Fix python 3.11 job and update python classifier in setup.cfg","message":"Fix python 3.11 job and update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nFixing the py311 unit test job.\n\nChange-Id: Ie53dc343927bc107a2445ed96a84d24d6513dc7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f749fad38861b41503361b70f8ca030e5c56ff16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/f749fad38861b41503361b70f8ca030e5c56ff16"}]},"branch":"refs/heads/master"},"26668baf7b9fa62dd0f0df57d6e1d1226510e39d":{"kind":"REWORK","_number":3,"created":"2024-01-22 20:14:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/904673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/murano","ref":"refs/changes/73/904673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/murano refs/changes/73/904673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/murano refs/changes/73/904673/3"}}},"commit":{"parents":[{"commit":"c898a310afbc27f12190446ef75d8b0bd12115eb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/c898a310afbc27f12190446ef75d8b0bd12115eb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:13:31.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-22 20:11:13.000000000","tz":-480},"subject":"Fix python 3.11 job and update python classifier in setup.cfg","message":"Fix python 3.11 job and update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nFixing the py311 unit test job:\n- Updating bindep.txt for mysql packages\n- Squash other two fixes in this commit to make job green\n  1. getargspec was deprecated in Python 3.0, and it\u0027s removed in\n     Python 3.11 in the favor of getfullargspec.\n  2. Opening files in U mode is removed in Python 3.11. This\n     fixes unit tests for me in Debian. Related Debian bug:\n     https://bugs.debian.org/1025013\n\nCo-Authored-By: Thomas Goirand \u003cthomas@goirand.fr\u003e\n\nChange-Id: Ie53dc343927bc107a2445ed96a84d24d6513dc7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/26668baf7b9fa62dd0f0df57d6e1d1226510e39d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/murano/commit/26668baf7b9fa62dd0f0df57d6e1d1226510e39d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
