)]}'
{"id":"openstack%2Fnetworking-bagpipe~1004443","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ibe4dbc4d560fcab447711557db7a431b3be2a706","project":"openstack/networking-bagpipe","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-14 19:10:10.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 15:45:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-14 14:11:42.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ibe4dbc4d560fcab447711557db7a431b3be2a706","subject":"Replace deprecated threading.Thread.setDaemon","status":"MERGED","created":"2026-09-07 15:42:57.000000000","updated":"2026-09-14 19:11:14.000000000","submitted":"2026-09-14 19:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004443","meta_rev_id":"dbf2e46d4dfc8575d1db8fd4bbe245424ddbf41d","_number":1004443,"virtual_id_number":1004443,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 19:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-09-14 09:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-14 14:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-09-14 14:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 17:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 09:07:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-09-14 09:07:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-14 09:07:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"430dc688f88f63d53343a8b884dc60e177bcb226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 15:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f0297a45458dbb6df022d7ff88d72396fecc12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 17:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5367a21c28114b7b8203ec884d5cabd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341a25b667924c878cee23b05d84401c : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bca713cb3efa4c84908d4bb88f3589ed : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/081670117c934691a05fb4487c4d5bfb : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/492caac600cb4b0784eacb80e58f5b3a : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e824f6e5c15f4389a00373951f04d1be : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/880bb8d222cc45e0bdde34c1a2c7523e : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0da872e810465488fcf513bfb9c756 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ee5f239e542d41d1b58fbbc3e1145dda : SUCCESS in 40m 34s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/84b3e5cba80d47ba8dd3574a05f60f80 : SUCCESS in 1h 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"23fad53237bdc80762c6ef836ad1456fc93c3b9a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-14 09:07:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f681a56511d56a76d4411026a2a7032b2540f951","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-14 14:11:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66503f5b018b2fda930761c06d313dd6880ae5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 14:12:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f12479e7d43ea5f9a0832e1ba9b72f20ef3e7bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 15:31:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc3e0e6e9a8246fba24f9ec07223c1b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658f093578a0401bbc259aed3e82185f : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11235b7f6d2f415dabca23f2289c7bff : SUCCESS in 8m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56a688fe9a594c66847eae373c04b9b0 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845fb2a2290541c3bd1648de59f75ebf : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ac0839e65a94a37b4dcadaea8e008c6 : SUCCESS in 5m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/22f2eec307fb4eed98a27e9559295606 : FAILURE in 17m 26s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/789aa97f3abb4aa88e152df162f7bf07 : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"25792e9756f190a4463f49e2c58b992d0b97f010","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 15:45:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50d15cb1477f9881fc7cf4cb27a9beaae319c63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:26:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ba122b0b1a4fc38e59d86724c0ec12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1f563c60b34367b3f50cf5f711b3e4 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc2dd566853d47db8e6d292342a358c3 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d513ed6ef8454cc5b75ac491eeb7c907 : SUCCESS in 6m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea865774a7b54b6e92d89f3eceb52c3b : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc06f2593ea41e78fa6c387585f5a9b : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/827f589c8e164559a865c91d72f86891 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82e4d90e33ba46b7bc047ed0b94175db : SUCCESS in 7m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4a3c437837a24e768cf378dcd6a59d14 : SUCCESS in 1h 28m 52s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ab416c45cc4a41f39727db64fd783590 : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0c4d8deb166d29221e81f0c75900541cf6b2da29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:27:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc457b61fd3ece9134b1e0935177d47f43131519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:10:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ced6c0bb98c644b0a6780f7b13da9f99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd3a9bed49542bdb78140a39ef21421 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63ec4e61e95747de87ef54601e6ff37a : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66edaf826246426f934c2228bd3d5939 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b12709260964938bdda685378023f1e : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e3ec84d9a944f4b871086800af81384 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/90f96ed068dc4ab79951442b320e0e7b : SUCCESS in 1h 29m 30s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/be2ade2dfee74e5b88c7d6e52f6d2d34 : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a62fcd9fe8246366fbd369068f9fdba1c5911f9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:10:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbf2e46d4dfc8575d1db8fd4bbe245424ddbf41d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 19:11:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c58b4511e954f7da7633928b5e46e91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e7b92cdb9a4959b2aff13fcce84bff : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36e0703968df61e026b436867122517e5680f6df","revisions":{"36e0703968df61e026b436867122517e5680f6df":{"kind":"REWORK","_number":1,"created":"2026-09-07 15:42:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/1004443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/43/1004443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/43/1004443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/43/1004443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/43/1004443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/43/1004443/1"}}},"commit":{"parents":[{"commit":"c5461f556dc3b203f1f34d71accfe6078e7b3ecc","subject":"hacking: enable hacking plugins from neutron-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5461f556dc3b203f1f34d71accfe6078e7b3ecc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:42:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:42:05.000000000","tz":540},"subject":"Replace deprecated threading.Thread.setDaemon","message":"Replace deprecated threading.Thread.setDaemon\n\nIt was deprecated in Python 3.10 in favor of the daemon property[1].\n\n[1] https://docs.python.org/3.10/library/threading.html#threading.Thread.setDaemon\n\nChange-Id: Ibe4dbc4d560fcab447711557db7a431b3be2a706\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/36e0703968df61e026b436867122517e5680f6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/36e0703968df61e026b436867122517e5680f6df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
