)]}'
{"id":"openstack%2Fnetworking-bagpipe~454739","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia66a1ace6f038b44523e033f232997f87204f7bf","project":"openstack/networking-bagpipe","branch":"master","topic":"agent-refactoring","hashtags":[],"change_id":"Ia66a1ace6f038b44523e033f232997f87204f7bf","subject":"Agent code refactoring","status":"MERGED","created":"2017-04-07 14:37:14.000000000","updated":"2017-05-17 15:43:55.000000000","submitted":"2017-05-17 15:43:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":69,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"310d0e22220e965a1bc5e5681859229710f48e79","_number":454739,"virtual_id_number":454739,"owner":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-05-17 15:43:55.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-05-17 14:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-05-17 14:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-17 14:50:07.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2017-05-17 15:43:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"dd9f942f2231d1a9bc18e93d01cce808ec824038","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-04-07 14:37:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16916081a5d330b5b92c5a39d2dbe829c0a7954d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-07 15:25:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/1/check/gate-networking-bagpipe-docs-ubuntu-xenial/ef1b9a7//doc/build/html/ : SUCCESS in 3m 03s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/1/check/gate-networking-bagpipe-pep8-ubuntu-xenial/ba93b77/ : FAILURE in 4m 52s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/1/check/gate-networking-bagpipe-python27-ubuntu-xenial/c9c1bd0/ : SUCCESS in 4m 42s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/1/check/gate-networking-bagpipe-python35/7c14ea6/ : SUCCESS in 4m 02s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/1/check/networking-bagpipe-coverage-ubuntu-xenial/50d9730/ : SUCCESS in 5m 11s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/1/check/gate-tempest-dsvm-networking-bagpipe/7f39046/ : SUCCESS in 46m 56s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/1/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/e3dace6/ : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b33deb2779590b3da3319d47edd105fa7e0082d4","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-04-11 14:21:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(40 comments)\n\nThe one \u0027big\u0027 thing is that there remain in net_info and port_info information that is already translated for use in a bagpipe-bgp API call, while I think we want there only mostly untransformed information from neutron service received over RPCs.","accounts_in_message":[],"_revision_number":1},{"id":"6e13fc013ea55c8c4b4bc5e3fb2637c83e91227f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-09 16:31:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aa7b3ea8059826313b0c36a889984e891487393","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-09 17:04:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/2/check/gate-networking-bagpipe-docs-ubuntu-xenial/6a46b17//doc/build/html/ : SUCCESS in 3m 04s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-pep8-ubuntu-xenial/4303292/ : FAILURE in 5m 39s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-python27-ubuntu-xenial/d872e3f/ : FAILURE in 4m 37s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-python35/8a51283/ : FAILURE in 4m 35s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/networking-bagpipe-coverage-ubuntu-xenial/2b3a2f1/ : FAILURE in 2m 53s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/2/check/gate-tempest-dsvm-networking-bagpipe/f357b92/ : FAILURE in 30m 04s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/bded212/ : FAILURE in 12m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d2ce4fb10a3f4257f17b7db80f486002df658c34","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-10 14:33:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(22 comments)\n\nThis really goes into the right direction.\n\nI still have a bunch of comments/suggestions/questions though...","accounts_in_message":[],"_revision_number":2},{"id":"62d8085e027a05da1eae0dbb1103184ac3ca494c","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-10 19:45:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46c15b0603cd23357c90117a16b8bbf740f25cf5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-10 20:50:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/2/check/gate-networking-bagpipe-docs-ubuntu-xenial/90c7785//doc/build/html/ : SUCCESS in 4m 05s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-pep8-ubuntu-xenial/8823fd4/ : FAILURE in 3m 00s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-python27-ubuntu-xenial/ea4152d/ : FAILURE in 2m 46s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/2/check/gate-networking-bagpipe-python35/71c992d/ : SUCCESS in 3m 52s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/networking-bagpipe-coverage-ubuntu-xenial/9524e4d/ : SUCCESS in 2m 52s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/2/check/gate-tempest-dsvm-networking-bagpipe/b564c44/ : SUCCESS in 1h 03m 58s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/2/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/2d4f160/ : FAILURE in 31m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8c9632f93150224eaa8af45224aec88e7df6bed9","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-10 22:05:27.000000000","message":"Patch Set 2:\n\nHey, weird to see unit tests fail with py27 and succeed with py35 !","accounts_in_message":[],"_revision_number":2},{"id":"5729558546ec0356b820b2801ce0cd3ac318a494","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-10 22:08:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67a28f0c16f991ce4818e761acb038138516577f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-15 08:32:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4aab9332fad235c836a56c1bdde5d157e079f3f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-15 09:22:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/3/check/gate-networking-bagpipe-docs-ubuntu-xenial/ee21310//doc/build/html/ : SUCCESS in 5m 30s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/3/check/gate-networking-bagpipe-pep8-ubuntu-xenial/11966c7/ : SUCCESS in 3m 11s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/3/check/gate-networking-bagpipe-python27-ubuntu-xenial/39521e8/ : SUCCESS in 2m 55s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/3/check/gate-networking-bagpipe-python35/c62e3c5/ : SUCCESS in 2m 40s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/3/check/networking-bagpipe-coverage-ubuntu-xenial/a735d13/ : FAILURE in 5m 03s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/3/check/gate-tempest-dsvm-networking-bagpipe/509aff2/ : SUCCESS in 49m 18s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/3/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/9695716/ : SUCCESS in 25m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9d04d2fe0d47d72a57e845a10e4a5f36657942ba","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-15 10:26:01.000000000","message":"Patch Set 3: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9143d1cd1557e2d875e0826a42bce89ff2ec4a21","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-15 10:54:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cdbbef80982184d701e79b9f83c4373850e5260","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-15 12:20:41.000000000","message":"Patch Set 3: Code-Review-1 -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"cbd934c9012af864f3fda0e449e1be07f6caa4c9","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-16 16:33:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c569cb64afee3f141391b8c9e325347269bfb75","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-16 17:34:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/4/check/gate-networking-bagpipe-docs-ubuntu-xenial/44fd2d3//doc/build/html/ : SUCCESS in 3m 52s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/4/check/gate-networking-bagpipe-pep8-ubuntu-xenial/afc8934/ : FAILURE in 5m 27s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/4/check/gate-networking-bagpipe-python27-ubuntu-xenial/1560dbd/ : FAILURE in 4m 35s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/4/check/gate-networking-bagpipe-python35/3f605a7/ : FAILURE in 2m 59s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/4/check/networking-bagpipe-coverage-ubuntu-xenial/c7a77e6/ : FAILURE in 3m 21s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/4/check/gate-tempest-dsvm-networking-bagpipe/658824b/ : SUCCESS in 58m 22s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/4/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/5be917a/ : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":4},{"id":"ebb719089bda80b86a2a29e1fb42ac186716a673","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-17 07:10:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"65b0707cd244190a7ee9cf9afd3cc3314469f16f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-17 07:11:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2e6b27a368de89f22cdebcf09ed9f2d193755022","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 08:12:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/5/check/gate-networking-bagpipe-docs-ubuntu-xenial/10352bf//doc/build/html/ : SUCCESS in 3m 54s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/5/check/gate-networking-bagpipe-pep8-ubuntu-xenial/9e903ca/ : FAILURE in 5m 02s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/5/check/gate-networking-bagpipe-python27-ubuntu-xenial/17b1902/ : SUCCESS in 3m 15s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/5/check/gate-networking-bagpipe-python35/b0ae193/ : SUCCESS in 3m 10s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/5/check/networking-bagpipe-coverage-ubuntu-xenial/1182d7f/ : SUCCESS in 3m 16s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/5/check/gate-tempest-dsvm-networking-bagpipe/9fd1305/ : SUCCESS in 57m 41s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/5/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/e958c0b/ : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":5},{"id":"67ca80a18c485e5e9bfa504f14877bea51233ce0","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-17 12:29:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ddfa51afe57b5daba263fd80f2bdec1aaff48716","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 13:24:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/6/check/gate-networking-bagpipe-docs-ubuntu-xenial/a9be674//doc/build/html/ : SUCCESS in 3m 57s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/6/check/gate-networking-bagpipe-pep8-ubuntu-xenial/96a6556/ : FAILURE in 5m 26s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/6/check/gate-networking-bagpipe-python27-ubuntu-xenial/0668d7f/ : SUCCESS in 3m 03s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/6/check/gate-networking-bagpipe-python35/9dfacdb/ : SUCCESS in 2m 47s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/6/check/networking-bagpipe-coverage-ubuntu-xenial/8ef7866/ : SUCCESS in 3m 56s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/6/check/gate-tempest-dsvm-networking-bagpipe/161efc3/ : SUCCESS in 53m 23s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/6/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/8db54d1/ : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":6},{"id":"edbce6984bc29997de9b756a803bb6acc8ecc73f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2017-05-17 13:24:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"25bf523ceaa978284a45f4375e2476ea134dc9dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 14:27:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/7/check/gate-networking-bagpipe-docs-ubuntu-xenial/aed09ca//doc/build/html/ : SUCCESS in 2m 49s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/7/check/gate-networking-bagpipe-pep8-ubuntu-xenial/ff3c55b/ : SUCCESS in 5m 29s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/7/check/gate-networking-bagpipe-python27-ubuntu-xenial/2ba203e/ : SUCCESS in 3m 11s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/7/check/gate-networking-bagpipe-python35/35335d8/ : SUCCESS in 4m 18s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/7/check/networking-bagpipe-coverage-ubuntu-xenial/bdc2526/ : SUCCESS in 5m 04s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/7/check/gate-tempest-dsvm-networking-bagpipe/a56e495/ : SUCCESS in 59m 48s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/7/check/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/1390e9c/ : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":7},{"id":"25151b08b863617602ac39f152e13c343040655f","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2017-05-17 14:50:07.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a806de7fa3d6100542c05175eaf2a2cc3629fc06","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 14:50:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"33dc61cd331df53f8e11b53ce2eac3a6fbe2489c","date":"2017-05-17 15:43:55.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":7},{"id":"f384d6dac44ff1e05b53dc2261aebb1a1e5e592e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 15:43:55.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-networking-bagpipe-docs-ubuntu-xenial http://docs-draft.openstack.org/39/454739/7/gate/gate-networking-bagpipe-docs-ubuntu-xenial/41df1cc//doc/build/html/ : SUCCESS in 5m 03s\n- gate-networking-bagpipe-pep8-ubuntu-xenial http://logs.openstack.org/39/454739/7/gate/gate-networking-bagpipe-pep8-ubuntu-xenial/50e1ba3/ : SUCCESS in 5m 46s\n- gate-networking-bagpipe-python27-ubuntu-xenial http://logs.openstack.org/39/454739/7/gate/gate-networking-bagpipe-python27-ubuntu-xenial/39a1f57/ : SUCCESS in 3m 28s\n- gate-networking-bagpipe-python35 http://logs.openstack.org/39/454739/7/gate/gate-networking-bagpipe-python35/566fca0/ : SUCCESS in 3m 27s\n- networking-bagpipe-coverage-ubuntu-xenial http://logs.openstack.org/39/454739/7/gate/networking-bagpipe-coverage-ubuntu-xenial/e041461/ : SUCCESS in 3m 54s\n- gate-tempest-dsvm-networking-bagpipe http://logs.openstack.org/39/454739/7/gate/gate-tempest-dsvm-networking-bagpipe/2a26916/ : SUCCESS in 53m 09s\n- gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial http://logs.openstack.org/39/454739/7/gate/gate-tempest-dsvm-networking-bgpvpn-bagpipe-ubuntu-xenial/e4d1af9/ : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6181a2e27caf383b82cf01bf050d2c4f5b656814","revisions":{"8b426d7b6840f9a23f84ed1f4ccd3d0e48fbf787":{"kind":"REWORK","_number":1,"created":"2017-04-07 14:37:14.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/1"}}},"commit":{"parents":[{"commit":"e7bbf23ec742fc4c521e47ee6a3708bbf4ba0cae","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e7bbf23ec742fc4c521e47ee6a3708bbf4ba0cae"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:35:06.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b426d7b6840f9a23f84ed1f4ccd3d0e48fbf787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b426d7b6840f9a23f84ed1f4ccd3d0e48fbf787"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e7bbf23ec742fc4c521e47ee6a3708bbf4ba0cae","is_merged_in_target_branch":true,"change_id":"Ia01b202905374eae5f25c8632482a060ef7e58d1","change_number":454498,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2c3c3602346c94700e925d2d4fd3b84447bec827":{"kind":"REWORK","_number":2,"created":"2017-05-09 16:31:45.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/2"}}},"commit":{"parents":[{"commit":"1d356400a6fa0f19dd43fa13615c2832634008c4","subject":"Merge \"devstack: fix gobgp installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1d356400a6fa0f19dd43fa13615c2832634008c4"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-09 16:30:43.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c3c3602346c94700e925d2d4fd3b84447bec827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c3c3602346c94700e925d2d4fd3b84447bec827"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d356400a6fa0f19dd43fa13615c2832634008c4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"563791821d29721a4a59a35d5db8c6ab3f931482":{"kind":"REWORK","_number":3,"created":"2017-05-15 08:32:23.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/3"}}},"commit":{"parents":[{"commit":"211b664dc62f17ecb29c4355b93fdce69fbefc27","subject":"devstack: set bagpipe-bgp rootwrap files even wo. bagpipe-bgp service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/211b664dc62f17ecb29c4355b93fdce69fbefc27"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-15 08:31:55.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/563791821d29721a4a59a35d5db8c6ab3f931482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/563791821d29721a4a59a35d5db8c6ab3f931482"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"211b664dc62f17ecb29c4355b93fdce69fbefc27","is_merged_in_target_branch":true,"change_id":"Ia00d6b5ccadc1de16d82b1b5cef1a20298b24bef","change_number":464207,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8d2da33b37794322271e41c7fadcd97390be6b6a":{"kind":"REWORK","_number":4,"created":"2017-05-16 16:33:49.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/4"}}},"commit":{"parents":[{"commit":"b67985402b8cea8830087d8666b3fff81c49c53c","subject":"bagpipe-bgp: properly cleanup vpn_instances on shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b67985402b8cea8830087d8666b3fff81c49c53c"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-16 16:32:41.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8d2da33b37794322271e41c7fadcd97390be6b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8d2da33b37794322271e41c7fadcd97390be6b6a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b67985402b8cea8830087d8666b3fff81c49c53c","is_merged_in_target_branch":true,"change_id":"I5d31d4ecfac69d53f932ae1bfb7f6dfc644b4657","change_number":464650,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a33e2ea36b1a4c1357c24e5609edb1271a23efc1":{"kind":"REWORK","_number":5,"created":"2017-05-17 07:11:45.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/5"}}},"commit":{"parents":[{"commit":"b67985402b8cea8830087d8666b3fff81c49c53c","subject":"bagpipe-bgp: properly cleanup vpn_instances on shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b67985402b8cea8830087d8666b3fff81c49c53c"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-17 07:11:27.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a33e2ea36b1a4c1357c24e5609edb1271a23efc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a33e2ea36b1a4c1357c24e5609edb1271a23efc1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b67985402b8cea8830087d8666b3fff81c49c53c","is_merged_in_target_branch":true,"change_id":"I5d31d4ecfac69d53f932ae1bfb7f6dfc644b4657","change_number":464650,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"412eba09e5f6462445375091fe1569dbcb1c5997":{"kind":"REWORK","_number":6,"created":"2017-05-17 12:29:18.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/6"}}},"commit":{"parents":[{"commit":"b67985402b8cea8830087d8666b3fff81c49c53c","subject":"bagpipe-bgp: properly cleanup vpn_instances on shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b67985402b8cea8830087d8666b3fff81c49c53c"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-17 12:29:00.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/412eba09e5f6462445375091fe1569dbcb1c5997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/412eba09e5f6462445375091fe1569dbcb1c5997"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b67985402b8cea8830087d8666b3fff81c49c53c","is_merged_in_target_branch":true,"change_id":"I5d31d4ecfac69d53f932ae1bfb7f6dfc644b4657","change_number":464650,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6181a2e27caf383b82cf01bf050d2c4f5b656814":{"kind":"REWORK","_number":7,"created":"2017-05-17 13:24:45.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/39/454739/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/39/454739/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/39/454739/7"}}},"commit":{"parents":[{"commit":"b67985402b8cea8830087d8666b3fff81c49c53c","subject":"bagpipe-bgp: properly cleanup vpn_instances on shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b67985402b8cea8830087d8666b3fff81c49c53c"}]}],"author":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-04-07 14:19:05.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2017-05-17 12:56:33.000000000","tz":120},"subject":"Agent code refactoring","message":"Agent code refactoring\n\nPermit to easily handle new service RPCs in agent\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ia66a1ace6f038b44523e033f232997f87204f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6181a2e27caf383b82cf01bf050d2c4f5b656814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6181a2e27caf383b82cf01bf050d2c4f5b656814"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b67985402b8cea8830087d8666b3fff81c49c53c","is_merged_in_target_branch":true,"change_id":"I5d31d4ecfac69d53f932ae1bfb7f6dfc644b4657","change_number":464650,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
