)]}'
{"id":"openstack%2Fnetworking-bagpipe~629174","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f","project":"openstack/networking-bagpipe","branch":"master","topic":"py3-pylint","hashtags":[],"change_id":"Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f","subject":"Upgrade pylint to a version that works with python3","status":"MERGED","created":"2019-01-08 13:11:19.000000000","updated":"2019-08-20 19:40:26.000000000","submitted":"2019-08-20 19:38:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"629174-1566329905691-66c16398","meta_rev_id":"b215a4a97eacf1c8ad5a2a36cdb51d845652e70e","_number":629174,"virtual_id_number":629174,"owner":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-20 15:59:18.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-20 19:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-08-19 21:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-08-19 21:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-14 14:55:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-08-14 14:51:09.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-08-19 21:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 16:54:02.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-14 14:51:09.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-14 14:55:16.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-19 21:22:56.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-19 21:25:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-20 15:59:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-08-20 19:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d93b548c2b0526c9e5cd9a073722e8f15f3cbd6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-01-08 13:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6358dc70db0746b802e90b45cee3e8ac356bc919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 14:34:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/629174/1/check/openstack-tox-pep8/8e599be/ : FAILURE in 8m 30s\n- openstack-tox-py27 http://logs.openstack.org/74/629174/1/check/openstack-tox-py27/83b85b2/ : SUCCESS in 7m 56s\n- openstack-tox-py35 http://logs.openstack.org/74/629174/1/check/openstack-tox-py35/d5de43a/ : SUCCESS in 8m 19s\n- openstack-tox-py36 http://logs.openstack.org/74/629174/1/check/openstack-tox-py36/180d5c1/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/74/629174/1/check/openstack-tox-docs/bc942fa/html/ : SUCCESS in 5m 23s\n- build-openstack-releasenotes http://logs.openstack.org/74/629174/1/check/build-openstack-releasenotes/39f3b21/html/ : SUCCESS in 4m 27s\n- requirements-check http://logs.openstack.org/74/629174/1/check/requirements-check/f9007fd/ : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/629174/1/check/openstack-tox-lower-constraints/185a54f/ : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/74/629174/1/check/openstack-tox-cover/8bf30f5/cover/ : SUCCESS in 8m 10s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/74/629174/1/check/legacy-tempest-dsvm-networking-bagpipe/566fc45/ : SUCCESS in 50m 35s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/74/629174/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/57c6c74/ : SUCCESS in 1h 11m 25s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/74/629174/1/check/legacy-networking-bagpipe-dsvm-fullstack/e1a0b67/ : FAILURE in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00ea5b3f4e6ee1e01a8e8ff45f3f6162e5a05379","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-01-09 07:25:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de2b6818566b22c80cf5d294ee976cf6740b8333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 07:51:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/629174/2/check/openstack-tox-pep8/273ac1e/ : FAILURE in 7m 14s\n- openstack-tox-py27 http://logs.openstack.org/74/629174/2/check/openstack-tox-py27/30ff16b/ : FAILURE in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/74/629174/2/check/openstack-tox-py35/62f4e5b/ : SUCCESS in 7m 39s\n- openstack-tox-py36 http://logs.openstack.org/74/629174/2/check/openstack-tox-py36/40361a6/ : SUCCESS in 7m 11s\n- openstack-tox-docs http://logs.openstack.org/74/629174/2/check/openstack-tox-docs/1252a2f/html/ : SUCCESS in 8m 00s\n- build-openstack-releasenotes http://logs.openstack.org/74/629174/2/check/build-openstack-releasenotes/2696c63/html/ : SUCCESS in 3m 47s\n- requirements-check http://logs.openstack.org/74/629174/2/check/requirements-check/8d6c898/ : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/629174/2/check/openstack-tox-lower-constraints/c59e8c5/ : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/74/629174/2/check/openstack-tox-cover/39a624b/ : FAILURE in 7m 22s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/74/629174/2/check/legacy-tempest-dsvm-networking-bagpipe/b65c34b/ : FAILURE in 15m 22s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/74/629174/2/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/d9dfc73/ : FAILURE in 19m 10s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/74/629174/2/check/legacy-networking-bagpipe-dsvm-fullstack/6ca9e37/ : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ccdbe04cce0099b9e4b256687b97820eb5903328","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-06-18 19:01:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f3b3783d3f5b549cd90f42602b54678fef889a73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:01:17.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"52c2a52d4d4c4de1deb295c33dd75097cb41c895","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-30 18:43:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2fe6fd172757c36c3cb49489e9bf2fddcb059aa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:23:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/629174/3/check/openstack-tox-pep8/f4f730a/ : FAILURE in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/74/629174/3/check/openstack-tox-py27/5d8fe06/ : FAILURE in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/74/629174/3/check/openstack-tox-py36/03e205f/ : SUCCESS in 6m 52s\n- openstack-tox-py37 http://logs.openstack.org/74/629174/3/check/openstack-tox-py37/c13e037/ : SUCCESS in 7m 54s\n- openstack-tox-docs http://logs.openstack.org/74/629174/3/check/openstack-tox-docs/8f1a315/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/74/629174/3/check/build-openstack-releasenotes/5180cfc/html/ : SUCCESS in 4m 27s\n- requirements-check http://logs.openstack.org/74/629174/3/check/requirements-check/4ea9928/ : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/629174/3/check/openstack-tox-lower-constraints/2ea89bf/ : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/74/629174/3/check/openstack-tox-cover/adb96ee/ : FAILURE in 3m 47s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/74/629174/3/check/legacy-tempest-dsvm-networking-bagpipe/f4e47d5/ : FAILURE in 21m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/74/629174/3/check/neutron-tempest-plugin-bgpvpn-bagpipe/02a3bfd/ : FAILURE in 20m 52s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/74/629174/3/check/legacy-networking-bagpipe-dsvm-fullstack/114c76d/ : FAILURE in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b72c5dc1f119649feb53a46004e63513b7e630b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-13 16:27:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8f61580f3fde647cd1ba8475d96b5d467f3ec85","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-13 16:54:02.000000000","message":"Patch Set 4:\n\nAfter debugging pep8 job, astroid version does not match pylint version.\nastroid 2.2.5 is too new for pylint 2.2.0. In the neutron job, we use astroid 2.1.0.","accounts_in_message":[],"_revision_number":4},{"id":"8902db7b65f24d451280ff0d034abdecb3211c8c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-13 17:23:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c514af99c80f8d2dceddb59c62da64b1663b0a66","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-13 17:25:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"99b57d359c3b8e8ad1812b51acd954dc37e50dc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 19:03:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eb511ab191492ebb36f15f591a2bfd : SUCCESS in 8m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5e23119d04942178c2ec246fd4b1531 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/494d0e8f7b30448894a27de47e8a01ca : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ce4fa38c7c84343b0f7a7a4091cd2b5 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3eaa86b7484d9583ef4f63b06edfa3 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dce7783ab4b493a90473d67f4ac991a : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22ab80d8c30446f5836d4dbc66210f49 : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e8a9d87008648b4ac7dbfdd99ab93b5 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e6e12fc407c4c4194880e2961f095ca : SUCCESS in 9m 00s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/8088ce8010224fc895887dc35a568d2f : SUCCESS in 1h 08m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3df387ba838e4fdc9aed2fe92ec6a090 : SUCCESS in 1h 36m 32s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3e9f4af983644d59a2fff4048989b9c1 : FAILURE in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a484c01b1dfb3bb7d840c63f9ad1703cd4da9872","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-14 14:51:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"03839f69304dfa83054803f3bd721968eaa59fd0","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 14:55:16.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ab3595ab9c4c24a4602d676e0de69a1d9c98f0aa","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-19 21:22:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"70f7d417ed076303e63c763c35f15917fecb7aa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:23:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"489564e57ed820b70fd10e6e16b643ddd0f4210c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-19 21:25:02.000000000","message":"Patch Set 6: Code-Review+1\n\npep8 test output looks OK","accounts_in_message":[],"_revision_number":6},{"id":"ae75df5d532d21a532848e553ff80cde2f24b52f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:01:53.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ecce5c87bc4aa5a577a01e27009340 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45743f43e2bf4a7f9c40239a651ceb59 : POST_FAILURE in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f4e382e027c4bfdb8094128e3c02243 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa834a79017846128cb4d15b55b56eac : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016d51c1cc044c7895738e8aed6fbb0e : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f49fd7e1486444e961aa2d2b0e65232 : SUCCESS in 4m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/999617273a784cb493c317d3e1326715 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/113e21235b5248f0bde40a92bde44cfa : SUCCESS in 1h 35m 39s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/757c26bf69c4422b8910ca87c79f1771 : SUCCESS in 1h 32m 11s","accounts_in_message":[],"_revision_number":6},{"id":"01de8ebfa12a7b91ae6d6258bb34574774274c37","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-08-20 15:59:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d548e9fc301aec5a2cd60b37652782bfb2f06d19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:57:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfecfb480204ae199e1da23df0884a1 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdd25ffae33e477c907e8c22912c0040 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18fa85e3c24a4fcf9cb0e8b310a019c8 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90d001e000024f31a9f587044a7134f2 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa28d077779f45c5b781784668364d5f : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/714f074fd3964e30b8495e9275f60ea6 : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3870a3bdcec44eeb28063d986693c97 : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/365567204aec48788bc6cfc08d75b142 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29bf67108f814b9aa3428d40bffc8ea2 : SUCCESS in 7m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dce34e1e540e4a6d8c8fdd5cc9184adc : SUCCESS in 1h 54m 46s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7688c0306e864c44ab09b420fecaa312 : FAILURE in 18m 39s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f48895dbd6ba422fba63f19b659ea0a9 : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":6},{"id":"81c04475fa849d9172c5564c1542818c70dd05e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:57:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e6193dab3d283f1098ff17de9a8adb202306aac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:38:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20c87b46bc14c5a81d3f31979186065 : SUCCESS in 10m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/929ef6ab74894afa8bd8b2ff67410146 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053ffb529ecf4cac8a4132fdbf3a83da : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2258ba9b86a14e7fb57d2fd42058171c : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e95f806473477994a7c7e7307ce7b8 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9539401e8a4e5b8e3dbfe97eedf513 : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e83003ea449d4988a5238af34b18e3c9 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8b7e605003184382b1bfcdb76d106899 : SUCCESS in 1h 37m 51s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3700360cb59949c49d88c7c92a901463 : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":6},{"id":"e82d1e382fdb307d13cfb12c40646621951dcd3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:38:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7af698ab764f86d1e96ede44840e7630f04baec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:40:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414313710c9c4b7c9ec538159dfca74c : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ba466abb8e2143ac6f14bce4f1832ae20b9fd56f","revisions":{"9cbc962bac4eef6442c0074b83c12fc74e762bba":{"kind":"REWORK","_number":1,"created":"2019-01-08 13:11:19.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/74/629174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/1"}}},"commit":{"parents":[{"commit":"5afc9f5c1f10887f094644399543d807b2ee0c82","subject":"agent extension: smoother handling of bagpipe-bgp unavailability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5afc9f5c1f10887f094644399543d807b2ee0c82"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:10:51.000000000","tz":120},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9cbc962bac4eef6442c0074b83c12fc74e762bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9cbc962bac4eef6442c0074b83c12fc74e762bba"}]},"branch":"refs/heads/master"},"861b7dd15ccc2a3f1e4174b3600eda5cc0fa7797":{"kind":"REWORK","_number":2,"created":"2019-01-09 07:25:38.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/74/629174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/2"}}},"commit":{"parents":[{"commit":"5afc9f5c1f10887f094644399543d807b2ee0c82","subject":"agent extension: smoother handling of bagpipe-bgp unavailability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5afc9f5c1f10887f094644399543d807b2ee0c82"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-09 07:25:26.000000000","tz":120},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/861b7dd15ccc2a3f1e4174b3600eda5cc0fa7797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/861b7dd15ccc2a3f1e4174b3600eda5cc0fa7797"}]},"branch":"refs/heads/master"},"0956900f3716be5182db43d45d08dce0ece076e9":{"kind":"REWORK","_number":3,"created":"2019-07-30 18:43:49.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/74/629174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/3"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-07-30 18:43:48.000000000","tz":-240},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0956900f3716be5182db43d45d08dce0ece076e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0956900f3716be5182db43d45d08dce0ece076e9"}]},"branch":"refs/heads/master"},"693f00a5b55c2519a1db7d9d8766c2b471c4f22c":{"kind":"REWORK","_number":4,"created":"2019-08-13 16:27:35.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/629174/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/4"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-13 16:27:16.000000000","tz":540},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/693f00a5b55c2519a1db7d9d8766c2b471c4f22c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/693f00a5b55c2519a1db7d9d8766c2b471c4f22c"}]},"branch":"refs/heads/master"},"7b93acd1428f7a2bcd56b195e374af9f739b29e7":{"kind":"REWORK","_number":5,"created":"2019-08-13 17:23:15.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/629174/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/5"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-13 17:21:29.000000000","tz":540},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nastroid version needs to be specified too.\nOtherwise, a newer version of astroid which conflicts with pylint 2.2.0\nwill be installed. astroid and pylint are tightly coupled,\nso both versions should be specified.\n\nurllib2 is only available for python2.\nIn python3, it has been merged into urllib.\nsix.moves.urllib needs to be used.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7b93acd1428f7a2bcd56b195e374af9f739b29e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7b93acd1428f7a2bcd56b195e374af9f739b29e7"}]},"branch":"refs/heads/master"},"ba466abb8e2143ac6f14bce4f1832ae20b9fd56f":{"kind":"REWORK","_number":6,"created":"2019-08-13 17:25:05.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/629174/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/629174/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/629174/6"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-01-08 13:09:38.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-13 17:24:58.000000000","tz":540},"subject":"Upgrade pylint to a version that works with python3","message":"Upgrade pylint to a version that works with python3\n\nFixes a few pylint warning from this recent pylint.\n\nlen-as-condition, inconsistent-return-statements and\nno-else-return rules are\nadded to the ignore list in .pylintrc because they are\npatterns commonly used in the neutron stadium and\nalready ignored in the neutron repo.\n\nastroid version needs to be specified too.\nOtherwise, a newer version of astroid which conflicts with pylint 2.2.0\nwill be installed. astroid and pylint are tightly coupled,\nso both versions should be specified.\n\nurllib2 is only available for python2.\nIn python3, it has been merged into urllib.\nsix.moves.urllib needs to be used.\n\nChange-Id: Ib8f25b8103f6bb429cf5fbaa7a071dab7326930f\nCo-Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ba466abb8e2143ac6f14bce4f1832ae20b9fd56f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ba466abb8e2143ac6f14bce4f1832ae20b9fd56f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
