)]}'
{"id":"openstack%2Fnetworking-bagpipe~636422","triplet_id":"openstack%2Fnetworking-bagpipe~master~I35c00d1e25e3c8abc4027219b0b3b3d34a480339","project":"openstack/networking-bagpipe","branch":"master","topic":"use-callback-payloads","hashtags":[],"change_id":"I35c00d1e25e3c8abc4027219b0b3b3d34a480339","subject":"use callback payloads for AGENT OVS_RESTARTED events","status":"MERGED","created":"2019-02-12 19:09:56.000000000","updated":"2019-10-10 15:48:29.000000000","submitted":"2019-10-10 15:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636422-1570722395331-4f81ad49","meta_rev_id":"0d2791850713f62ea425ded59c1fffa1111cac6b","_number":636422,"virtual_id_number":636422,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:46:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-10-10 12:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-10-10 12:45:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 20:39:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-10-07 07:15:04.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-10-10 12:45:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-10 15:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12dbd249e2ea8c02d6af80642b712fa074593f94","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-12 19:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49b63935c558afa70788bd4972d4d275dd6e17a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-12 19:56:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/1/check/openstack-tox-pep8/1207ba0/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/1/check/openstack-tox-py27/f5ff506/ : SUCCESS in 8m 06s\n- openstack-tox-py35 http://logs.openstack.org/22/636422/1/check/openstack-tox-py35/bce13a2/ : SUCCESS in 7m 24s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/1/check/openstack-tox-py36/45ef764/ : SUCCESS in 8m 27s\n- openstack-tox-docs http://logs.openstack.org/22/636422/1/check/openstack-tox-docs/a5034a2/html/ : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/1/check/openstack-tox-lower-constraints/26be9d9/ : FAILURE in 7m 40s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/1/check/openstack-tox-cover/2988c52/cover/ : SUCCESS in 7m 37s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/1/check/legacy-tempest-dsvm-networking-bagpipe/c48f213/ : FAILURE in 30m 12s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/22/636422/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/18e7bed/ : FAILURE in 25m 34s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/1/check/legacy-networking-bagpipe-dsvm-fullstack/a96a820/ : FAILURE in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c8261af4e235c0dce060d8223f5b786522d70dd","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-14 13:19:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44b61819ff17956969e3676732f19a1c62a72d7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 14:04:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/1/check/openstack-tox-pep8/5c5935f/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/1/check/openstack-tox-py27/11ba023/ : SUCCESS in 7m 33s\n- openstack-tox-py35 http://logs.openstack.org/22/636422/1/check/openstack-tox-py35/8f651d0/ : SUCCESS in 8m 29s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/1/check/openstack-tox-py36/855d785/ : SUCCESS in 7m 17s\n- openstack-tox-docs http://logs.openstack.org/22/636422/1/check/openstack-tox-docs/3afb20e/html/ : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/1/check/openstack-tox-lower-constraints/601ad2e/ : FAILURE in 8m 33s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/1/check/openstack-tox-cover/aea1725/cover/ : SUCCESS in 8m 01s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/1/check/legacy-tempest-dsvm-networking-bagpipe/d1debf0/ : FAILURE in 26m 15s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/22/636422/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/3fa947a/ : FAILURE in 28m 23s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/1/check/legacy-networking-bagpipe-dsvm-fullstack/c7954b3/ : FAILURE in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0205300264b8bbf35d27d7c5967d40ede239f9e","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-14 14:06:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"30434b195515267f9ec21511f8b247dfcbab96b6","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-14 14:06:39.000000000","message":"Topic changed from fix-lowerconstraints to use-callback-payloads","accounts_in_message":[],"_revision_number":2},{"id":"3def6fe629739124e469402640dc7a5f52f9442a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 15:03:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/check/openstack-tox-pep8/7d381e5/ : SUCCESS in 7m 37s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/a4b2cbd/ : SUCCESS in 8m 00s\n- openstack-tox-py35 http://logs.openstack.org/22/636422/2/check/openstack-tox-py35/2a0c49a/ : SUCCESS in 8m 05s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/check/openstack-tox-py36/f4f6654/ : SUCCESS in 8m 02s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/check/openstack-tox-docs/62d3d7b/html/ : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/2/check/openstack-tox-lower-constraints/9c728fa/ : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/2/check/openstack-tox-cover/82cdcc8/cover/ : SUCCESS in 12m 36s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bagpipe/e35a8b8/ : FAILURE in 27m 36s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/145e8de/ : FAILURE in 25m 45s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/2/check/legacy-networking-bagpipe-dsvm-fullstack/3a9062a/ : FAILURE in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b022f17c0ed083d8b19866ab7c919cc3ffdca244","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-26 21:36:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7c8e567047340e12d81fe1042404569df6afbd77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 23:09:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/check/openstack-tox-pep8/adbd3a1/ : SUCCESS in 8m 04s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/95e015b/ : SUCCESS in 9m 21s\n- openstack-tox-py35 http://logs.openstack.org/22/636422/2/check/openstack-tox-py35/3eb4d9f/ : SUCCESS in 8m 56s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/check/openstack-tox-py36/4408741/ : SUCCESS in 7m 12s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/check/openstack-tox-docs/d5762d0/html/ : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/2/check/openstack-tox-lower-constraints/defabe8/ : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/2/check/openstack-tox-cover/399707e/cover/ : SUCCESS in 9m 40s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bagpipe/1124a3d/ : SUCCESS in 57m 46s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/75bcd34/ : SUCCESS in 1h 28m 37s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/2/check/legacy-networking-bagpipe-dsvm-fullstack/9953cbb/ : FAILURE in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cedfa9528a431f67ba1b70e757a1ba441ffc39fc","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-27 07:26:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec9b340fac5fa31ce674af004825e73c0c154119","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-27 11:18:27.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"4fe851a35c5d5287638b9472883c46f8ae59ba6e","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-27 11:18:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84b9049b2bdb96fab0d3b79f65de42b8bceb6efb","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-06-19 19:38:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c1a7c24d6121867aacd131ad9a59975380ff33ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:38:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21465658cba516ef68585fca92764a30250d4016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:22:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/gate/openstack-tox-pep8/5db1d11/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py27/e7c6bc4/ : SUCCESS in 7m 47s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py36/e7b6039/ : SUCCESS in 8m 06s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py37/d3bd18d/ : SUCCESS in 8m 51s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/gate/openstack-tox-docs/3201f60/html/ : SUCCESS in 6m 48s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/gate/legacy-tempest-dsvm-networking-bagpipe/8af6ebc/ : SUCCESS in 48m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/42d16c0/ : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6a1034cd26014b31bad8cff7c86ac12f7669c7f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:51:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/check/openstack-tox-pep8/ff181e0/ : SUCCESS in 8m 20s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/af60a6b/ : SUCCESS in 7m 42s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/check/openstack-tox-py36/b5c0059/ : SUCCESS in 7m 12s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/check/openstack-tox-py37/d442a3f/ : SUCCESS in 10m 04s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/check/openstack-tox-docs/8384cbe/html/ : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/2/check/openstack-tox-lower-constraints/40ecf90/ : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/2/check/openstack-tox-cover/b827dbe/cover/ : SUCCESS in 8m 32s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bagpipe/dfddc4d/ : FAILURE in 11m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/f28275d/ : FAILURE in 2h 04m 27s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/2/check/legacy-networking-bagpipe-dsvm-fullstack/1da4d00/ : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85c03ed787e468485d9bbce246b1a8dad89cf502","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-06-24 19:09:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e7857f0d6c859c2b555e9c0ae60d27f3f8a6e52c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 21:04:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/check/openstack-tox-pep8/75e27ec/ : RETRY_LIMIT in 2m 14s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/0d26f90/ : RETRY_LIMIT in 2m 07s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/check/openstack-tox-py36/fc5bcc4/ : RETRY_LIMIT in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/check/openstack-tox-py37/e54ec05/ : RETRY_LIMIT in 2m 43s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/check/openstack-tox-docs/a5c3ed4/html/ : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/2/check/openstack-tox-lower-constraints/099ff65/ : RETRY_LIMIT in 2m 27s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/2/check/openstack-tox-cover/e499079/ : RETRY_LIMIT in 2m 24s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bagpipe/03fe91c/ : SUCCESS in 1h 10m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/64b234b/ : SUCCESS in 1h 46m 17s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/2/check/legacy-networking-bagpipe-dsvm-fullstack/51becb6/ : FAILURE in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c71a101897f0cd4c1e02fb66da2d587078a2b02","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 20:39:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f9807e51eef3faa9869968d7ffb5665947da1d66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 22:21:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/check/openstack-tox-pep8/7384428/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/aa0da77/ : SUCCESS in 6m 15s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/check/openstack-tox-py36/9536345/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/check/openstack-tox-py37/d543c1a/ : SUCCESS in 6m 50s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/check/openstack-tox-docs/0e226ed/html/ : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/636422/2/check/openstack-tox-lower-constraints/a480252/ : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/22/636422/2/check/openstack-tox-cover/6a130c3/cover/ : SUCCESS in 7m 23s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/check/legacy-tempest-dsvm-networking-bagpipe/76c5cac/ : SUCCESS in 55m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/511f083/ : SUCCESS in 1h 36m 57s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/22/636422/2/check/legacy-networking-bagpipe-dsvm-fullstack/79f0e00/ : FAILURE in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e9e8457d887d6ffe42251b904374b180a731819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 22:21:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ceea0b77dd6f5cf9d7b63c92a84cbc86b3446cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 00:14:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/gate/openstack-tox-pep8/2dae818/ : SUCCESS in 7m 17s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py27/2df6528/ : SUCCESS in 7m 40s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py36/92c6137/ : SUCCESS in 6m 58s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py37/a193468/ : SUCCESS in 7m 29s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/gate/openstack-tox-docs/68bb4d9/html/ : SUCCESS in 5m 37s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/gate/legacy-tempest-dsvm-networking-bagpipe/5514ebd/ : SUCCESS in 50m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/be7c899/ : SUCCESS in 1h 49m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7659a8b3e89982a157838400ef1d468be11e3acb","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-06-29 14:52:44.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"1fe685313b5e52e9bd63c18ee77b28a667168225","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-06-29 14:52:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3bffb0703bc2b0d4ee45ad62f6075ec9fa6c9b3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 14:53:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9b3c28ffa0aef93ca59dfc5eb9145f390641824","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 16:45:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/gate/openstack-tox-pep8/ad9e2b4/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py27/0fa0369/ : SUCCESS in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py36/3896901/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py37/788eefd/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/gate/openstack-tox-docs/bf7906c/html/ : SUCCESS in 4m 06s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/gate/legacy-tempest-dsvm-networking-bagpipe/ffc4821/ : SUCCESS in 52m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/a6dbb95/ : SUCCESS in 1h 47m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a8d244eacf63d43943c723babc5dbf610c16e3d6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-07-18 14:08:42.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"a83c7009b5b64502ad639796d65ec75ff95272c6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-07-18 14:08:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20e8451e4b04156d2c5d3836af2c9636a54ce560","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:09:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5440af6892aa4ebab8b408607637ca2f3548a62e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:53:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/636422/2/gate/openstack-tox-pep8/f581b34/ : SUCCESS in 6m 48s\n- openstack-tox-py27 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py27/e7cf0cc/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py36/89153e9/ : SUCCESS in 7m 02s\n- openstack-tox-py37 http://logs.openstack.org/22/636422/2/gate/openstack-tox-py37/09eb990/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/22/636422/2/gate/openstack-tox-docs/77836b2/html/ : SUCCESS in 7m 15s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/22/636422/2/gate/legacy-tempest-dsvm-networking-bagpipe/d812652/ : SUCCESS in 53m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/22/636422/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/4b6b445/ : SUCCESS in 1h 42m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e666db9434013ca969d50ee01c9213d1f0de4d15","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-03 19:42:59.000000000","message":"Removed Code-Review+2 by Thomas Morin \u003ctmmorin.orange@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"d0f10d833f6bc81f67e2d3537c3b044b406e8e73","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-03 19:43:01.000000000","message":"Removed Workflow+1 by Thomas Morin \u003ctmmorin.orange@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"30d3da4f15ee3e442ed8d177b1784dce4bf8e528","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-03 19:43:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5a832bdd90e0d0bb35d2c15836da364a0a33a6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:10:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac38a7dffc634526810c8dd5cdf9cb48 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd8071608ab407c946b6225941836b8 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52bb02eafac440208de681586f1ee746 : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3527036cea64b47bc91ba56c4297d47 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed2129a871e489e85be4e8186696fe4 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593c3504acd84c65bec36946753a5a0b : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8837d197e4244cf918348637ddec6a9 : SUCCESS in 8m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/be726df71a3340abbabef26284a23f8f : SUCCESS in 1h 25m 49s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/977dee77a2f043d5afd824d3298cf1ab : FAILURE in 18m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/425c3aaf7dac4d92aecb8bbca842fc0b : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5b714d9ebc4de20b1a1dc557158c2de6232eef9e","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-07 07:15:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"671b64643000593a84046da335afdabf307886f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 07:15:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4d1d3107ede40f4e30b80d667c3e0661b403f47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 09:01:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85d1625ac4ac4e069863d06c0f884085 : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a371748db9264f2aaa94d1b90a96245f : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcf2df7d7014e7fa8c259267d2e6906 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95a140e0a2464bf78e4c3d0a975b62a4 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e77ffad5b8a41688f941afc8e685ea5 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9d0531afc7524aa6aac2079e1ad2ba4c : SUCCESS in 1h 42m 00s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4dfe6fe6d8414f9eb558c67d173173ad : SUCCESS in 1h 29m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7193f66ea8d5a5dd2d6814511a4d625989ae7149","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-10 12:39:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2acea8c99df68787ec5e3444761cfabb624f4063","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-10 12:45:43.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"77fad225c446ecb91f2fa0d83eef2315fda44027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:12:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6399a859b8a0455eb3a0e32559223063 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3afce8bd5de54dfb8e00c2bd6612dfa5 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f900783d9c4c4db3b3241ac24bcaa5 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fedae92ad32c4a3ba34dae00226e5c00 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f666e9f8f5a74c3685dcdd8ca0271616 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7f3401197d54693b86d529ce97870dc : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/714626ae1dbe44f6adadeb560de6a2b2 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f7f75ac414f140819704b53a9649296c : SUCCESS in 1h 19m 14s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6e9858b01abd430fa2d806b2676ca0c5 : FAILURE in 24m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/65c501b302da420ebc4f093071189f74 : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":3},{"id":"fb35ffdccdc57e87e183a58a27837b90093a68f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:12:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"079def00c381344e25ed936dae457bcbdb91567d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:46:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6632ea2f1f224636bf4f636c5e6e299c : SUCCESS in 8m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e07096deab3a4d5694496649265fdb6e : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aed9f0351c63495d8c4dc15d73ad7b30 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9594494cd2cb4a40ab18c2b192c85787 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3b952859f7479faee981c41bfa5ff9 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b0acd3acfc634ad8992d114dccbe9e7c : SUCCESS in 1h 29m 07s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4a81e1f76af348a197b1de219414ebfd : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":3},{"id":"efdc513be7ce39d8f81b9f554976c425d906da3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:46:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0d2791850713f62ea425ded59c1fffa1111cac6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:48:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202814f6b8574b27964e7e1b25d47c1b : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1fabbc010fe0e14ea69da94e5e2cfd94bf72bd00","revisions":{"3600d85486cead41d520731c28d2033b90664997":{"kind":"REWORK","_number":1,"created":"2019-02-12 19:09:56.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/22/636422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/22/636422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/1"}}},"commit":{"parents":[{"commit":"8e7613ce40eb688557f79dd7b771c796100802cf","subject":"stop using common db mixin methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8e7613ce40eb688557f79dd7b771c796100802cf"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-12 19:09:46.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-12 19:09:46.000000000","tz":-420},"subject":"use callback payloads for AGENT OVS_RESTARTED events","message":"use callback payloads for AGENT OVS_RESTARTED events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor AGENT OVS_RESTARTED events. In addition it updates the unit tests to\nalso use payloads for OVS_RESTARTED based events as needed.\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nDepends-On: https://review.openstack.org/#/c/636421/\nChange-Id: I35c00d1e25e3c8abc4027219b0b3b3d34a480339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3600d85486cead41d520731c28d2033b90664997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3600d85486cead41d520731c28d2033b90664997"}]},"branch":"refs/heads/master"},"35079d930971cbf8a51fd314ba13ab1faa38df36":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-14 14:06:26.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/22/636422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/22/636422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/2"}}},"commit":{"parents":[{"commit":"c91af7b573545a4a59a85b2bc6e09f3fae12a504","subject":"fix lower constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c91af7b573545a4a59a85b2bc6e09f3fae12a504"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-12 19:09:46.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-14 14:06:08.000000000","tz":-420},"subject":"use callback payloads for AGENT OVS_RESTARTED events","message":"use callback payloads for AGENT OVS_RESTARTED events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor AGENT OVS_RESTARTED events. In addition it updates the unit tests to\nalso use payloads for OVS_RESTARTED based events as needed.\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nDepends-On: https://review.openstack.org/#/c/636421/\nChange-Id: I35c00d1e25e3c8abc4027219b0b3b3d34a480339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/35079d930971cbf8a51fd314ba13ab1faa38df36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/35079d930971cbf8a51fd314ba13ab1faa38df36"}]},"branch":"refs/heads/master"},"1fabbc010fe0e14ea69da94e5e2cfd94bf72bd00":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-10 12:39:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/22/636422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/22/636422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/22/636422/3"}}},"commit":{"parents":[{"commit":"a59b5c75751f5acf4e30223851ca1e6323c62e9a","subject":"Merge \"Force copy of dict items in \"for\" loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a59b5c75751f5acf4e30223851ca1e6323c62e9a"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-12 19:09:46.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-10 12:39:51.000000000","tz":0},"subject":"use callback payloads for AGENT OVS_RESTARTED events","message":"use callback payloads for AGENT OVS_RESTARTED events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor AGENT OVS_RESTARTED events. In addition it updates the unit tests to\nalso use payloads for OVS_RESTARTED based events as needed.\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nDepends-On: https://review.openstack.org/#/c/636421/\nChange-Id: I35c00d1e25e3c8abc4027219b0b3b3d34a480339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1fabbc010fe0e14ea69da94e5e2cfd94bf72bd00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1fabbc010fe0e14ea69da94e5e2cfd94bf72bd00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
