)]}'
{"id":"openstack%2Fnetworking-bagpipe~641057","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7e2b8737703060f532463bae823f99f9ecadc0df","project":"openstack/networking-bagpipe","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I7e2b8737703060f532463bae823f99f9ecadc0df","subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","status":"MERGED","created":"2019-03-05 15:50:07.000000000","updated":"2019-08-16 22:47:04.000000000","submitted":"2019-08-16 22:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"641057-1565995624391-16970a7d","meta_rev_id":"9d0126e441bea64e020112735055878ce9ca5b6f","_number":641057,"virtual_id_number":641057,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2019-08-16 19:35:50.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-16 22:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-08-16 05:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-08-16 08:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-08-16 08:20:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 12:09:50.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-16 05:29:17.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-08-16 08:20:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-16 22:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb82167298007da7403b7552e49363246c0525fd","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-03-05 15:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f244915683e231c35b7da267b458998f765fb0a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 15:50:21.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027required_projects\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: networking-bagpipe-tempest\n      parent: tempest-full-py3\n      timeout: 10800\n      required_projects:\n        - openstack-infra/devstack-gate\n        - openstack/tempest\n        - openstack/neutron\n        - openstack/networking-bagpipe\n      vars:\n        tempest_concurrency: 4\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 104, column 3","accounts_in_message":[],"_revision_number":1},{"id":"29aaf3e6441ea2c51271b1729048e6656448b12f","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-09 16:03:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e117ee5f30ab5b54d47847c3199048ac6e2fd5b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:04:11.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  Unknown project openstack-infra/devstack-gate\n\nThe error appears in the following job stanza:\n\n  job:\n      name: networking-bagpipe-tempest\n      parent: tempest-full-py3\n      timeout: 10800\n      required-projects:\n        - openstack-infra/devstack-gate\n        - openstack/tempest\n        - openstack/neutron\n        - openstack/networking-bagpipe\n      vars:\n        tempest_concurrency: 4\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 87, column 3","accounts_in_message":[],"_revision_number":2},{"id":"6db723025f92c3bcaf64e409e4021d569bad47ef","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-10 13:57:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45ca77ee9dbe3440381066251f1e7ebd9be3b81d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:51:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/57/641057/3/check/openstack-tox-pep8/0bd7aa2/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/57/641057/3/check/openstack-tox-py27/97f0aa4/ : SUCCESS in 7m 21s\n- openstack-tox-py36 http://logs.openstack.org/57/641057/3/check/openstack-tox-py36/8404840/ : SUCCESS in 8m 10s\n- openstack-tox-py37 http://logs.openstack.org/57/641057/3/check/openstack-tox-py37/70e5df2/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/57/641057/3/check/openstack-tox-docs/7ff93a3/html/ : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/641057/3/check/openstack-tox-lower-constraints/595304f/ : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/57/641057/3/check/openstack-tox-cover/6c7b509/cover/ : SUCCESS in 9m 59s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/57/641057/3/check/legacy-tempest-dsvm-networking-bagpipe/bedfd89/ : SUCCESS in 1h 01m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/57/641057/3/check/neutron-tempest-plugin-bgpvpn-bagpipe/7e6c913/ : SUCCESS in 1h 41m 34s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/57/641057/3/check/legacy-networking-bagpipe-dsvm-fullstack/8dff0d2/ : FAILURE in 11m 14s (non-voting)\n- networking-bagpipe-tempest http://logs.openstack.org/57/641057/3/check/networking-bagpipe-tempest/c24d1e0/ : SUCCESS in 1h 46m 35s (non-voting)\n- networking-bagpipe-bgpvpn-tempest http://logs.openstack.org/57/641057/3/check/networking-bagpipe-bgpvpn-tempest/036b249/ : SUCCESS in 1h 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"157ecefa19c5e5a95742b2dc067a9409413e7112","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-17 07:38:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d7014378e8214b9645127775b4e210eec9299783","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-17 20:43:21.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e1c0f8211573ee54c6e48591bc28f688537c6686","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-07-27 06:46:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nA small -1 because of the pending question.\n\nBut thanks guys for taking care of networking-bagpipe!","accounts_in_message":[],"_revision_number":3},{"id":"3276298c75c22c845def20244e7573a9f1764f51","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-30 18:58:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d17f480a016f6fd4e4dc362f1687bd3c42f0439","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:58:37.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  Job networking-bagpipe-bgpvpn-tempest not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python3-train-jobs-neutron\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - check-requirements\n        - periodic-stable-jobs-neutron\n      check:\n        jobs:\n          - openstack-tox-lower-constraints:\n              # Currently, we cannot specify that we want neutron\n              # checked out from master using tox-siblings for this\n              # job, so this always fails.\n              voting: false\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-networking-bagpipe-dsvm-fullstack:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              voting: false\n          - networking-bagpipe-tempest\n          - networking-bagpipe-bgpvpn-tempest\n      gate:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - networking-bagpipe-bgpvpn-tempest\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"aedc96e9a22edcd3cee9399935789cf5e0d23d68","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-15 11:34:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0370dbba4e7bb445e67cc6df7a6b57629cc6a923","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:34:18.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  Job networking-bagpipe-bgpvpn-tempest not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python3-train-jobs-neutron\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - check-requirements\n        - periodic-stable-jobs-neutron\n      check:\n        jobs:\n          - openstack-tox-lower-constraints:\n              # Currently, we cannot specify that we want neutron\n              # checked out from master using tox-siblings for this\n              # job, so this always fails.\n              voting: false\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-networking-bagpipe-dsvm-fullstack:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              voting: false\n          - networking-bagpipe-tempest\n          - networking-bagpipe-bgpvpn-tempest\n      gate:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - networking-bagpipe-bgpvpn-tempest\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"2a0a74f3e7ed97fdb72619db32ae22a5b9954510","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-15 12:01:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"33338efafaf8a08d4f226d70c2d87787a6396851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:02:08.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  Job networking-bagpipe-bgpvpn-tempest not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python3-train-jobs-neutron\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - check-requirements\n        - periodic-stable-jobs-neutron\n      check:\n        jobs:\n          - openstack-tox-lower-constraints:\n              # Currently, we cannot specify that we want neutron\n              # checked out from master using tox-siblings for this\n              # job, so this always fails.\n              voting: false\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-networking-bagpipe-dsvm-fullstack:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              voting: false\n          - networking-bagpipe-tempest\n          - networking-bagpipe-bgpvpn-tempest:\n              required-projects:\n                - openstack/neutron-tempest-plugin\n      gate:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - openstack-tox-py37:\n              required-projects:\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n          - legacy-tempest-dsvm-networking-bagpipe:\n              required-projects:\n                - openstack/neutron\n                - openstack/networking-bgpvpn\n                - openstack/networking-sfc\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - networking-bagpipe-bgpvpn-tempest:\n              required-projects:\n                - openstack/neutron-tempest-plugin\n\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":5},{"id":"881a679c9d0fd88f4284c8a3b11d774c439fa7a5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-15 12:09:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"584db74c1bc36c042940705ac6e850569e5c8bee","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-15 12:18:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fac6071668946648e903efa1b461c84bcda5caab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 14:07:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e2ded244a0425ab1265210097283e9 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a99f000dfeb4c3ca50e2b6aae8f53dc : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42256807cb8245eeb9dd1b721a22dadb : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7ea20875fe0478993c099f8a2ff4530 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03da27bdacd44e6bad6d6d868a408cf8 : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/565a00516da740f8849491963060dbe1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bdd04e875494788a449c6d2cb6ee826 : SUCCESS in 6m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ad861b88ccf047379ddf4e9259406672 : SUCCESS in 1h 47m 01s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d4d7d14def764c688cb848e4f9ab6f22 : FAILURE in 12m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/26a3875411414533b1d885fcb1cf044e : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":6},{"id":"455a9a2a8fb4c9d11364b9eb726210a4b94b2e32","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-08-16 05:29:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"874540775c2a82dba0b87ef0dc7000769a4d7ea1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-16 08:20:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"af8a650ad5d9268e23cf4fde25853ce04235917d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 08:20:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d85d8714582efb4bb7eb281dfa3b07490b95f3a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:52:15.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3554ded279549a0b3a64c803e50c1ed : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6939d07fd614fe6a22ce7e6752da835 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d16cde40eb684d8397f620b9f5dc11af : SUCCESS in 8m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/131fe8ea827e4254aac508c92af7ef29 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9286fb1fc9b54814b5bc327e14da1970 : FAILURE in 5m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3fcc4db1ed224ca593cf6276ae61f97c : SUCCESS in 1h 28m 27s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/45fdcb828de04041b884cfed5b46d428 : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":6},{"id":"df5bebe640a85461553c597b35288993f24b02bc","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-16 19:35:43.000000000","message":"Patch Set 6:\n\nLooks like tox-docs failed with an unrelated error:\n\n  File \"/home/zuul/src/opendev.org/openstack/networking-bagpipe/.tox/docs/lib/python3.6/site-packages/neutron/plugins/ml2/drivers/linuxbridge/agent/arp_protect.py\", line 20, in \u003cmodule\u003e\n    import tenacity\nModuleNotFoundError: No module named \u0027tenacity\u0027","accounts_in_message":[],"_revision_number":6},{"id":"4e3e2dfdb50e0a2cb4f5e8807914b077de526887","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-16 19:35:50.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f4e9b0b0eb71287bef943121ad5d761dbda7155d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 21:02:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157dc838b82f484f85579278739a0336 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/402ce7377daf418d95f843308c6fb686 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771187ab413746e8a6f18f7de0bb94b5 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96bf370953124e749d26b8aac5bce83e : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fc9c892aa04a0bb7c21f21b3dcccfb : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c88dbf24163646e2b8a9f0cee04e55cd : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6af7488a89740f68e4aee52fba95a79 : SUCCESS in 6m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e30c22e993544472b8b9b18aa5f8e235 : SUCCESS in 1h 25m 35s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e15a8fc7db724007b99f88d997d7dd60 : FAILURE in 13m 09s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1ff0b98295d249b2a3f2419404c38575 : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":6},{"id":"b7539982a51ce1a476578b190fa7325c128ca9c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 21:02:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5ef2742dadcdca6f8cba3ddf5273d5fb09d59733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:47:04.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17674183e2c34df78076d3cd8cbd9add : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4126e39c0ee74ea19c20c788ed3de83a : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fac2b09f5d584e52b462e15199039606 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eba5c05da4e34f1d90d21ba81927cd17 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61df36810d7f4b48819f2a38040a7c49 : SUCCESS in 6m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d7c08f399ec547a9836738a78ec06794 : SUCCESS in 1h 38m 48s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/35299dd7413242089ae4245a11551226 : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":6},{"id":"d6b1677c1af63a9ec9ab2988e7090503302fe989","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:47:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cec4ba74eb114b4fec30f9a981cfd37daeec6988","revisions":{"5cbc7fe66f7387fa5b986b434905f30eed40b341":{"kind":"REWORK","_number":1,"created":"2019-03-05 15:50:07.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/57/641057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/1"}}},"commit":{"parents":[{"commit":"983ca46c3db42cf172ba0b42305032f6a496722d","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/983ca46c3db42cf172ba0b42305032f6a496722d"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:49:59.000000000","tz":-300},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5cbc7fe66f7387fa5b986b434905f30eed40b341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5cbc7fe66f7387fa5b986b434905f30eed40b341"}]},"branch":"refs/heads/master"},"b4839f1ca6b096cc030410b6ac26113a2c516570":{"kind":"REWORK","_number":2,"created":"2019-07-09 16:03:57.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/57/641057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/2"}}},"commit":{"parents":[{"commit":"983ca46c3db42cf172ba0b42305032f6a496722d","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/983ca46c3db42cf172ba0b42305032f6a496722d"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-07-09 16:03:51.000000000","tz":-240},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4839f1ca6b096cc030410b6ac26113a2c516570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4839f1ca6b096cc030410b6ac26113a2c516570"}]},"branch":"refs/heads/master"},"39050bcb0c720e3e002cf4c4fbefdca688d0c907":{"kind":"REWORK","_number":3,"created":"2019-07-10 13:57:07.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/57/641057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/3"}}},"commit":{"parents":[{"commit":"983ca46c3db42cf172ba0b42305032f6a496722d","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/983ca46c3db42cf172ba0b42305032f6a496722d"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-07-10 13:57:01.000000000","tz":-240},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39050bcb0c720e3e002cf4c4fbefdca688d0c907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39050bcb0c720e3e002cf4c4fbefdca688d0c907"}]},"branch":"refs/heads/master"},"921695206bddd63a06fa5115d814fb006b3b7e18":{"kind":"REWORK","_number":4,"created":"2019-07-30 18:58:24.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/57/641057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/4"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-07-30 18:58:13.000000000","tz":-240},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/921695206bddd63a06fa5115d814fb006b3b7e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/921695206bddd63a06fa5115d814fb006b3b7e18"}]},"branch":"refs/heads/master"},"3822a2c49238750d6a2f839c74e7bb438f4ac5d0":{"kind":"REWORK","_number":5,"created":"2019-08-15 12:01:54.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/57/641057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/5"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-15 12:01:43.000000000","tz":540},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3822a2c49238750d6a2f839c74e7bb438f4ac5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3822a2c49238750d6a2f839c74e7bb438f4ac5d0"}]},"branch":"refs/heads/master"},"cec4ba74eb114b4fec30f9a981cfd37daeec6988":{"kind":"REWORK","_number":6,"created":"2019-08-15 12:18:33.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/57/641057/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/641057/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/641057/6"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-05 15:47:32.000000000","tz":-300},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-15 12:18:10.000000000","tz":540},"subject":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3","message":"Migrate bagpipe legacy tempest jobs to zuulv3 and py3\n\nIntroduce new zuul v3 versions of the old\nlegacy-tempest-dsvm-networking-bagpipe and\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe jobs, for modernization\nand compliance with the python 3 goal.\n\nChange-Id: I7e2b8737703060f532463bae823f99f9ecadc0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cec4ba74eb114b4fec30f9a981cfd37daeec6988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cec4ba74eb114b4fec30f9a981cfd37daeec6988"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
