)]}'
{"id":"openstack%2Fnetworking-bagpipe~655298","triplet_id":"openstack%2Fnetworking-bagpipe~master~I1bf4bee2d427548253b9a96115f4a751a21b3b45","project":"openstack/networking-bagpipe","branch":"master","topic":"opendev","hashtags":[],"change_id":"I1bf4bee2d427548253b9a96115f4a751a21b3b45","subject":"Use opendev repository","status":"MERGED","created":"2019-04-24 05:40:26.000000000","updated":"2019-05-21 09:26:28.000000000","submitted":"2019-05-21 09:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655298-1558430788009-f9d4f00a","meta_rev_id":"d7dfce917af34eb933f5e55274db2d13cc3df2b2","_number":655298,"virtual_id_number":655298,"owner":{"_account_id":26584,"name":"hou le","email":"houl7@chinaunicom.cn","username":"houl7"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-05-21 09:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-05-21 07:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-05-17 21:13:13.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2019-05-21 07:47:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 21:13:13.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-21 07:47:06.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-05-21 09:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71b99d87328285992d918e308c23de7f971431bf","author":{"_account_id":26584,"name":"hou le","email":"houl7@chinaunicom.cn","username":"houl7"},"date":"2019-04-24 05:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a689db6b63bcdf5d5265bee80a863b1ef18e87bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 06:59:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/655298/1/check/test-release-openstack/5fdd33d/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/98/655298/1/check/openstack-tox-pep8/77f9ee2/ : SUCCESS in 7m 39s\n- openstack-tox-py27 http://logs.openstack.org/98/655298/1/check/openstack-tox-py27/7efe06f/ : SUCCESS in 9m 56s\n- openstack-tox-py35 http://logs.openstack.org/98/655298/1/check/openstack-tox-py35/5f7c683/ : SUCCESS in 7m 42s\n- openstack-tox-py36 http://logs.openstack.org/98/655298/1/check/openstack-tox-py36/9790e44/ : SUCCESS in 7m 10s\n- openstack-tox-py37 http://logs.openstack.org/98/655298/1/check/openstack-tox-py37/95858f4/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/98/655298/1/check/openstack-tox-docs/aabf749/html/ : SUCCESS in 7m 41s\n- build-openstack-releasenotes http://logs.openstack.org/98/655298/1/check/build-openstack-releasenotes/d87580d/html/ : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/655298/1/check/openstack-tox-lower-constraints/5d75d2f/ : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/98/655298/1/check/openstack-tox-cover/e916771/cover/ : SUCCESS in 7m 47s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/98/655298/1/check/legacy-tempest-dsvm-networking-bagpipe/28876e9/ : SUCCESS in 49m 49s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/98/655298/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/fd68c7f/ : SUCCESS in 1h 17m 43s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/98/655298/1/check/legacy-networking-bagpipe-dsvm-fullstack/691873a/ : FAILURE in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2231d8123e2cc726ee77e15708088298848ec8f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-17 21:13:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f69e6abdb0c166f097d18e245f0be891e8293a6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-05-21 07:47:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14ca27d3974f52501412e2fbbb397984e120dc6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 07:47:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ba6cd4ff1773f59367067c0c2f00c725ba929a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:26:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/655298/1/gate/test-release-openstack/01b8bc1/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/98/655298/1/gate/openstack-tox-pep8/1c0eacc/ : SUCCESS in 7m 45s\n- openstack-tox-py27 http://logs.openstack.org/98/655298/1/gate/openstack-tox-py27/b5e5c31/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/98/655298/1/gate/openstack-tox-py36/7868235/ : SUCCESS in 8m 58s\n- openstack-tox-py37 http://logs.openstack.org/98/655298/1/gate/openstack-tox-py37/cea7f18/ : SUCCESS in 8m 24s\n- openstack-tox-docs http://logs.openstack.org/98/655298/1/gate/openstack-tox-docs/980ba61/html/ : SUCCESS in 6m 32s\n- build-openstack-releasenotes http://logs.openstack.org/98/655298/1/gate/build-openstack-releasenotes/7281591/html/ : SUCCESS in 4m 48s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/98/655298/1/gate/legacy-tempest-dsvm-networking-bagpipe/75bb896/ : SUCCESS in 56m 23s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/98/655298/1/gate/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/77626fc/ : SUCCESS in 1h 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d210e6147390b585f128c1c865cf62a10dc22fae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:26:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4700056a36b53547e00fd61f5a16285137b7f366","revisions":{"4700056a36b53547e00fd61f5a16285137b7f366":{"kind":"REWORK","_number":1,"created":"2019-04-24 05:40:26.000000000","uploader":{"_account_id":26584,"name":"hou le","email":"houl7@chinaunicom.cn","username":"houl7"},"ref":"refs/changes/98/655298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/655298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/655298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/655298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/655298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/655298/1"}}},"commit":{"parents":[{"commit":"782370e6839c82f8cfaf31d7b4c8c107e37efac0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/782370e6839c82f8cfaf31d7b4c8c107e37efac0"}]}],"author":{"name":"Le Hou","email":"houl7@chinaunicom.cn","date":"2019-04-24 05:39:28.000000000","tz":480},"committer":{"name":"Le Hou","email":"houl7@chinaunicom.cn","date":"2019-04-24 05:40:21.000000000","tz":480},"subject":"Use opendev repository","message":"Use opendev repository\n\nupdate README and tox.in\n\nChange-Id: I1bf4bee2d427548253b9a96115f4a751a21b3b45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4700056a36b53547e00fd61f5a16285137b7f366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4700056a36b53547e00fd61f5a16285137b7f366"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
