)]}'
{"id":"openstack%2Fnetworking-bagpipe~659881","triplet_id":"openstack%2Fnetworking-bagpipe~master~I4d06c893a69fcd8d1eba40afd94992b17155bdd4","project":"openstack/networking-bagpipe","branch":"master","topic":"bp/neutronlib-decouple-db","hashtags":[],"change_id":"I4d06c893a69fcd8d1eba40afd94992b17155bdd4","subject":"use AuthenticIPNetwork from neutron-lib","status":"MERGED","created":"2019-05-17 21:15:27.000000000","updated":"2019-06-18 00:51:19.000000000","submitted":"2019-06-18 00:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659881-1560819079545-018dc456","meta_rev_id":"7253edbc9bc492151253fe4a9f9a1793d115d2ad","_number":659881,"virtual_id_number":659881,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-06-14 17:23:48.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-06-18 00:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-06-10 22:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-14 14:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-06-14 14:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-17 21:07:26.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-06-14 14:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":1,"date":"2019-06-17 21:07:26.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 22:00:45.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-14 14:22:36.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 14:56:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-17 21:07:26.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-06-18 00:51:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdb4faac9e51e17a4e05c0a490631d0913795b57","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-17 21:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"024e17ff6d3190c7f08a831d5db3f806c59d51a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:53:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/1/check/openstack-tox-pep8/abf4505/ : SUCCESS in 8m 41s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/1/check/openstack-tox-py27/fdbef32/ : SUCCESS in 8m 44s\n- openstack-tox-py35 http://logs.openstack.org/81/659881/1/check/openstack-tox-py35/b8ef203/ : SUCCESS in 8m 26s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/1/check/openstack-tox-py36/2288fad/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/1/check/openstack-tox-py37/a4d5048/ : SUCCESS in 7m 01s\n- openstack-tox-docs http://logs.openstack.org/81/659881/1/check/openstack-tox-docs/4eacd73/html/ : SUCCESS in 4m 52s\n- requirements-check http://logs.openstack.org/81/659881/1/check/requirements-check/40fec13/ : FAILURE in 3m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/659881/1/check/openstack-tox-lower-constraints/1593363/ : FAILURE in 10m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/81/659881/1/check/openstack-tox-cover/910e27e/cover/ : SUCCESS in 7m 30s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/1/check/legacy-tempest-dsvm-networking-bagpipe/55e5f9c/ : SUCCESS in 57m 22s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/81/659881/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/d295c3a/ : SUCCESS in 1h 34m 29s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/81/659881/1/check/legacy-networking-bagpipe-dsvm-fullstack/735017b/ : FAILURE in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a68f036fdd581874fdfa9b605f892905ef6df0d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-22 15:59:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3374abebf7542c2bac8eb11090122d62b29aba44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 17:28:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/2/check/openstack-tox-pep8/4288fea/ : SUCCESS in 8m 13s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/2/check/openstack-tox-py27/821ad84/ : SUCCESS in 9m 37s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/2/check/openstack-tox-py36/cf019a0/ : SUCCESS in 8m 50s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/2/check/openstack-tox-py37/8feb626/ : SUCCESS in 9m 51s\n- openstack-tox-docs http://logs.openstack.org/81/659881/2/check/openstack-tox-docs/9c48148/html/ : SUCCESS in 7m 59s\n- requirements-check http://logs.openstack.org/81/659881/2/check/requirements-check/d2b1b96/ : FAILURE in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/659881/2/check/openstack-tox-lower-constraints/295e610/ : FAILURE in 6m 14s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/81/659881/2/check/openstack-tox-cover/29ab8e3/cover/ : SUCCESS in 14m 06s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/2/check/legacy-tempest-dsvm-networking-bagpipe/9095b56/ : FAILURE in 54m 05s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/81/659881/2/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/fe90755/ : SUCCESS in 1h 19m 27s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/81/659881/2/check/legacy-networking-bagpipe-dsvm-fullstack/62fd69e/ : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2eb686a030f6b31cb939426d2827a872c7c9d9a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-06 21:33:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6db80a3c774f03d56f97973e3bce8a2bb2dd20d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-06 21:43:29.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"09aa70d5e3a90506da2851e7f5acb87cf5d98e35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:34:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/4/check/openstack-tox-pep8/9cace50/ : SUCCESS in 7m 24s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/4/check/openstack-tox-py27/c09b898/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/4/check/openstack-tox-py36/668f2a1/ : SUCCESS in 6m 53s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/4/check/openstack-tox-py37/a77390e/ : SUCCESS in 7m 22s\n- openstack-tox-docs http://logs.openstack.org/81/659881/4/check/openstack-tox-docs/b78417d/html/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/659881/4/check/openstack-tox-lower-constraints/d71cb10/ : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/81/659881/4/check/openstack-tox-cover/87792fa/cover/ : SUCCESS in 8m 29s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/4/check/legacy-tempest-dsvm-networking-bagpipe/5949e61/ : SUCCESS in 1h 03m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/81/659881/4/check/neutron-tempest-plugin-bgpvpn-bagpipe/0c2393b/ : SUCCESS in 1h 49m 28s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/81/659881/4/check/legacy-networking-bagpipe-dsvm-fullstack/c1254e5/ : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ea5a6aaa3d42306d4ecd12bedabb72811f5ed8a7","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-06-07 16:48:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7361d8c02739ed407fb7b798d57a1db01646445f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-10 22:00:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"455ab7c5522142916ea8aa6e20377c07822694d7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-11 14:13:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"511578abc666371aa7221e16c16fe568cd18e0f3","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-14 13:22:40.000000000","message":"Patch Set 4:\n\nThis patch has been out for almost a month.\nCan we pls try to land it so we can get the neutron patch in too?","accounts_in_message":[],"_revision_number":4},{"id":"e5b7a2436a1ea8b92f0a4aa4ed8ca5a717c7a22d","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:22:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bb5dd2bf30958eafea59a5eecb05916e60504ac4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-14 14:56:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7db90648e4d5f94ba9a943755209b92cccbf2afc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:56:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bc27d542b6bb656706f8859407d342040261df39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:55:54.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/4/gate/openstack-tox-pep8/eb21dbd/ : SUCCESS in 8m 32s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py27/dd14922/ : SUCCESS in 7m 54s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py36/005ae38/ : SUCCESS in 8m 38s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py37/d79fdc5/ : SUCCESS in 7m 50s\n- openstack-tox-docs http://logs.openstack.org/81/659881/4/gate/openstack-tox-docs/dc32f99/html/ : SUCCESS in 6m 16s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/4/gate/legacy-tempest-dsvm-networking-bagpipe/c42c60e/ : SUCCESS in 56m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/81/659881/4/gate/neutron-tempest-plugin-bgpvpn-bagpipe/8b16f7c/ : FAILURE in 1h 51m 09s","accounts_in_message":[],"_revision_number":4},{"id":"bd29b85878b1a600592e5965201012804ad7a59a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-14 17:23:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1b831c9b32eba7bdac77315966f3cf0eee55bf33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 19:18:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/4/check/openstack-tox-pep8/8ea8a63/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/4/check/openstack-tox-py27/da5971d/ : SUCCESS in 8m 39s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/4/check/openstack-tox-py36/a33fd87/ : FAILURE in 7m 43s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/4/check/openstack-tox-py37/0755b70/ : SUCCESS in 10m 09s\n- openstack-tox-docs http://logs.openstack.org/81/659881/4/check/openstack-tox-docs/2da8f24/html/ : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/659881/4/check/openstack-tox-lower-constraints/22a2452/ : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/81/659881/4/check/openstack-tox-cover/92490df/cover/ : SUCCESS in 8m 01s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/4/check/legacy-tempest-dsvm-networking-bagpipe/1a093f7/ : SUCCESS in 58m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/81/659881/4/check/neutron-tempest-plugin-bgpvpn-bagpipe/36b4a44/ : SUCCESS in 1h 47m 29s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/81/659881/4/check/legacy-networking-bagpipe-dsvm-fullstack/e21b5ec/ : FAILURE in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"109eee4687d9fd6b965408679dc48aae6d607e28","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-06-17 21:07:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a68973909f15ad3f4a57aab955ea1117f93aef60","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-06-17 21:07:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fd2b48a86c3e0266b4983a78caf0c3638c0315e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:58:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/4/check/openstack-tox-pep8/d24cea0/ : SUCCESS in 9m 30s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/4/check/openstack-tox-py27/e2b2bad/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/4/check/openstack-tox-py36/5879378/ : SUCCESS in 10m 36s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/4/check/openstack-tox-py37/2c2bc58/ : SUCCESS in 7m 46s\n- openstack-tox-docs http://logs.openstack.org/81/659881/4/check/openstack-tox-docs/e67d376/html/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/659881/4/check/openstack-tox-lower-constraints/acad032/ : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/81/659881/4/check/openstack-tox-cover/1321669/cover/ : SUCCESS in 8m 22s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/4/check/legacy-tempest-dsvm-networking-bagpipe/ad5425e/ : SUCCESS in 1h 00m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/81/659881/4/check/neutron-tempest-plugin-bgpvpn-bagpipe/466734d/ : SUCCESS in 1h 46m 07s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/81/659881/4/check/legacy-networking-bagpipe-dsvm-fullstack/e3fe7a6/ : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eaa4026ffdba18bbafbfab4f5f75c9f4261ee5d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:59:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"62c72af38f089dfcd348b73cd68e1aa55f6f8df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:51:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"86dee36b2e9a53feb06987acac393cae8ba2eca0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:51:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/81/659881/4/gate/openstack-tox-pep8/0dc0fa9/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py27/bfd6273/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py36/024eb83/ : SUCCESS in 7m 46s\n- openstack-tox-py37 http://logs.openstack.org/81/659881/4/gate/openstack-tox-py37/3bd7071/ : SUCCESS in 8m 02s\n- openstack-tox-docs http://logs.openstack.org/81/659881/4/gate/openstack-tox-docs/2875538/html/ : SUCCESS in 6m 44s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/81/659881/4/gate/legacy-tempest-dsvm-networking-bagpipe/7897e91/ : SUCCESS in 1h 11m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/81/659881/4/gate/neutron-tempest-plugin-bgpvpn-bagpipe/6bb3630/ : SUCCESS in 1h 45m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b7d1816bc66256c31bd3e5459927a05cb2a8c58a","revisions":{"7cdf0f921b34e4c754e402232f322f1f8621c5ca":{"kind":"REWORK","_number":1,"created":"2019-05-17 21:15:27.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/81/659881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/81/659881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/1"}}},"commit":{"parents":[{"commit":"782370e6839c82f8cfaf31d7b4c8c107e37efac0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/782370e6839c82f8cfaf31d7b4c8c107e37efac0"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-17 21:15:11.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-17 21:15:27.000000000","tz":-360},"subject":"use AuthenticIPNetwork from neutron-lib","message":"use AuthenticIPNetwork from neutron-lib\n\nThis patch:\n- Switches from using neutron for AuthenticIPNetwork to use neutron-lib.\n- Bumps the required neutron-lib version to 1.26.0 to get the latest.\n\nChange-Id: I4d06c893a69fcd8d1eba40afd94992b17155bdd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cdf0f921b34e4c754e402232f322f1f8621c5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cdf0f921b34e4c754e402232f322f1f8621c5ca"}]},"branch":"refs/heads/master"},"7c773797ed635b26389147fb1892d5564d5c5b0e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-22 15:59:17.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/81/659881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/81/659881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/2"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-17 21:15:11.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 15:59:11.000000000","tz":-360},"subject":"use AuthenticIPNetwork from neutron-lib","message":"use AuthenticIPNetwork from neutron-lib\n\nThis patch:\n- Switches from using neutron for AuthenticIPNetwork to use neutron-lib.\n- Bumps the required neutron-lib version to 1.26.0 to get the latest.\n\nChange-Id: I4d06c893a69fcd8d1eba40afd94992b17155bdd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7c773797ed635b26389147fb1892d5564d5c5b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7c773797ed635b26389147fb1892d5564d5c5b0e"}]},"branch":"refs/heads/master"},"c560a04ceba65a6f6084201dd4fa018861d7f2c7":{"kind":"REWORK","_number":3,"created":"2019-06-06 21:33:19.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/81/659881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/81/659881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/3"}}},"commit":{"parents":[{"commit":"16151fcee4b5d79fd9937391dd9636fa160218c3","subject":"Update lower-constraints to valid with recent changes in Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/16151fcee4b5d79fd9937391dd9636fa160218c3"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-17 21:15:11.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-06 21:23:11.000000000","tz":-360},"subject":"use AuthenticIPNetwork from neutron-lib","message":"use AuthenticIPNetwork from neutron-lib\n\nThis patch:\n- Switches from using neutron for AuthenticIPNetwork to use neutron-lib.\n- Bumps the required neutron-lib version to 1.26.0 to get the latest.\n\nChange-Id: I4d06c893a69fcd8d1eba40afd94992b17155bdd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c560a04ceba65a6f6084201dd4fa018861d7f2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c560a04ceba65a6f6084201dd4fa018861d7f2c7"}]},"branch":"refs/heads/master"},"b7d1816bc66256c31bd3e5459927a05cb2a8c58a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-06 21:43:29.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/81/659881/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/81/659881/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/81/659881/4"}}},"commit":{"parents":[{"commit":"16151fcee4b5d79fd9937391dd9636fa160218c3","subject":"Update lower-constraints to valid with recent changes in Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/16151fcee4b5d79fd9937391dd9636fa160218c3"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-17 21:15:11.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-06 21:43:22.000000000","tz":-360},"subject":"use AuthenticIPNetwork from neutron-lib","message":"use AuthenticIPNetwork from neutron-lib\n\nThis patch:\n- Switches from using neutron for AuthenticIPNetwork to use neutron-lib.\n\nChange-Id: I4d06c893a69fcd8d1eba40afd94992b17155bdd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b7d1816bc66256c31bd3e5459927a05cb2a8c58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b7d1816bc66256c31bd3e5459927a05cb2a8c58a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
