)]}'
{"id":"openstack%2Fnetworking-bagpipe~662192","triplet_id":"openstack%2Fnetworking-bagpipe~master~I46cdfc3592014995674c0060d43cfd0ae0a29693","project":"openstack/networking-bagpipe","branch":"master","topic":"test-gate","hashtags":[],"change_id":"I46cdfc3592014995674c0060d43cfd0ae0a29693","subject":"DNM: test gate","status":"ABANDONED","created":"2019-05-30 12:18:51.000000000","updated":"2019-06-06 21:21:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c3eb25e992393e54f9e5ed82d80d6c1e0f14116","_number":662192,"virtual_id_number":662192,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-30 16:54:52.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":-1,"date":"2019-05-30 13:34:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 13:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95c6a84e56de19ac59e19b848a14082e2cd10142","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-30 12:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793e7c20e34bf96e352bb418607f42a65bf5f601","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 13:34:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/662192/1/check/openstack-tox-pep8/8d2e0a9/ : FAILURE in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/92/662192/1/check/openstack-tox-py27/5c38881/ : FAILURE in 7m 33s\n- openstack-tox-py36 http://logs.openstack.org/92/662192/1/check/openstack-tox-py36/88962e9/ : FAILURE in 8m 10s\n- openstack-tox-py37 http://logs.openstack.org/92/662192/1/check/openstack-tox-py37/1ab3dfd/ : FAILURE in 6m 41s\n- openstack-tox-docs http://logs.openstack.org/92/662192/1/check/openstack-tox-docs/c594627/html/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/662192/1/check/openstack-tox-lower-constraints/cd2308c/ : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/92/662192/1/check/openstack-tox-cover/7b54753/ : FAILURE in 5m 50s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/92/662192/1/check/legacy-tempest-dsvm-networking-bagpipe/8952cf4/ : SUCCESS in 1h 02m 37s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/92/662192/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/046d008/ : FAILURE in 27m 40s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/92/662192/1/check/legacy-networking-bagpipe-dsvm-fullstack/39a4d72/ : FAILURE in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11311ed86decb84347b119c36c77557af57a2549","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-30 16:54:52.000000000","message":"Patch Set 1:\n\nCreated bug https://bugs.launchpad.net/networking-bagpipe/+bug/1831127","accounts_in_message":[],"_revision_number":1},{"id":"f3e91d6e35eff0ede49deebbaebc2fc47d5273a0","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-06 21:21:38.000000000","message":"Abandoned\n\ngate is fixed now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc897c1cb341d835ef23d8070a1bd16807935c98","revisions":{"cc897c1cb341d835ef23d8070a1bd16807935c98":{"kind":"REWORK","_number":1,"created":"2019-05-30 12:18:51.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/92/662192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/92/662192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/92/662192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/92/662192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/92/662192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/92/662192/1"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-30 12:18:45.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-30 12:18:45.000000000","tz":-360},"subject":"DNM: test gate","message":"DNM: test gate\n\nDO NOT MERGE\n\nThis just a dummy patch to test the gate jobs.\n\nChange-Id: I46cdfc3592014995674c0060d43cfd0ae0a29693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cc897c1cb341d835ef23d8070a1bd16807935c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cc897c1cb341d835ef23d8070a1bd16807935c98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
