)]}'
{"id":"openstack%2Fnetworking-bagpipe~662305","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7d571386b29234c627f973bfac2005a5577e54ed","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1831127","hashtags":[],"change_id":"I7d571386b29234c627f973bfac2005a5577e54ed","subject":"Switch to native openflow implementation","status":"MERGED","created":"2019-05-30 19:55:05.000000000","updated":"2019-06-06 13:37:46.000000000","submitted":"2019-06-06 13:37:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662305-1559828266853-4b762db2","meta_rev_id":"ded4e02203db998d2681d04764d6513be8d1eec5","_number":662305,"virtual_id_number":662305,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"date":"2019-06-06 09:44:18.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-06-06 13:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-06-05 21:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-05-31 13:54:00.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-05-31 13:55:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-05-31 14:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2019-06-05 21:41:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 13:54:00.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-31 13:55:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-31 14:01:02.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-06-05 21:41:55.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-06 13:37:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa34a32128b23dcf311491b4edb75714545bfd18","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-30 19:55:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2565767246487a1f5cb7bb5bea55d30730636913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:54:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/1/check/openstack-tox-pep8/591f6e8/ : SUCCESS in 7m 11s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/1/check/openstack-tox-py27/003ccc9/ : SUCCESS in 7m 26s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/1/check/openstack-tox-py36/0ce8859/ : SUCCESS in 7m 42s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/1/check/openstack-tox-py37/a57ec2d/ : SUCCESS in 7m 29s\n- openstack-tox-docs http://logs.openstack.org/05/662305/1/check/openstack-tox-docs/bf68e07/html/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/662305/1/check/openstack-tox-lower-constraints/e962f76/ : FAILURE in 6m 08s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/662305/1/check/openstack-tox-cover/1ae40cb/cover/ : SUCCESS in 7m 42s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/1/check/legacy-tempest-dsvm-networking-bagpipe/8e8d071/ : SUCCESS in 52m 47s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/05/662305/1/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/0585799/ : FAILURE in 29m 22s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/05/662305/1/check/legacy-networking-bagpipe-dsvm-fullstack/66dd28f/ : FAILURE in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1897577fd0e913b28eb4f46ad92875b2fd432b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-30 21:33:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"39d2175742091da4c0ca5e4abee365689f3aeb8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 22:28:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/2/check/openstack-tox-pep8/927d94b/ : SUCCESS in 9m 45s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/2/check/openstack-tox-py27/287245f/ : SUCCESS in 9m 38s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/2/check/openstack-tox-py36/dc94137/ : SUCCESS in 9m 08s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/2/check/openstack-tox-py37/258f953/ : SUCCESS in 7m 13s\n- openstack-tox-docs http://logs.openstack.org/05/662305/2/check/openstack-tox-docs/0daf168/html/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/662305/2/check/openstack-tox-lower-constraints/02bc427/ : FAILURE in 5m 44s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/662305/2/check/openstack-tox-cover/f8a03d7/cover/ : SUCCESS in 7m 09s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/2/check/legacy-tempest-dsvm-networking-bagpipe/cdd3df0/ : SUCCESS in 49m 33s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe http://logs.openstack.org/05/662305/2/check/legacy-tempest-dsvm-networking-bgpvpn-bagpipe/cd10daf/ : FAILURE in 28m 11s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/05/662305/2/check/legacy-networking-bagpipe-dsvm-fullstack/fa62c70/ : FAILURE in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"681a051f363c86c85680c218c7bd5799ddcfe22e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 08:34:27.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"46a2afb988ddcfbc06f0bbc9851b62f129301619","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:12:36.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f3e18dcb71bc9fa8022104080c55c70b0eeb41f7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:59:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"47637263bf83fd8530781d933be61265cdf68b56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:01:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/5/check/openstack-tox-pep8/348b744/ : SUCCESS in 12m 44s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/5/check/openstack-tox-py27/fcff4e2/ : SUCCESS in 8m 13s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/5/check/openstack-tox-py36/0a7115a/ : SUCCESS in 7m 37s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/5/check/openstack-tox-py37/da76570/ : SUCCESS in 7m 37s\n- openstack-tox-docs http://logs.openstack.org/05/662305/5/check/openstack-tox-docs/f0a6d90/html/ : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/662305/5/check/openstack-tox-lower-constraints/d687de7/ : FAILURE in 6m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/662305/5/check/openstack-tox-cover/179eea7/cover/ : SUCCESS in 8m 53s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/5/check/legacy-tempest-dsvm-networking-bagpipe/31439b8/ : SUCCESS in 57m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/05/662305/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/5bca010/ : SUCCESS in 1h 57m 48s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/05/662305/5/check/legacy-networking-bagpipe-dsvm-fullstack/13d5ec7/ : FAILURE in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a13a0f3d1002b606f6f22a1685ef2f4f9eeca94","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-31 13:54:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ad037f80186eb655c00973f923f2960cfc523fa9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-31 13:55:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1c997c6aff91f9b46900f49816ce8bc1a21da105","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-31 14:01:02.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM.  Long term though we should start figuring out how to move these integration points to neutron-lib.","accounts_in_message":[],"_revision_number":5},{"id":"b4373156d8d7d002e6a9d03f48600c92da60db0a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-05 21:41:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7576635254d20f56dafefdfd6f6f6bbed4a4e5cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:42:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"66362ff4c2801d9c926a52a662ea986b4b97a823","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:42:03.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/5/gate/openstack-tox-pep8/26a026b/ : SUCCESS in 7m 57s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py27/6a78d6d/ : SUCCESS in 7m 17s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py36/bcff06d/ : SUCCESS in 8m 13s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py37/ee0c92e/ : SUCCESS in 10m 05s\n- openstack-tox-docs http://logs.openstack.org/05/662305/5/gate/openstack-tox-docs/55833f8/html/ : SUCCESS in 5m 22s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/5/gate/legacy-tempest-dsvm-networking-bagpipe/e686f8a/ : FAILURE in 1h 14m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/05/662305/5/gate/neutron-tempest-plugin-bgpvpn-bagpipe/82c6dea/ : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":5},{"id":"46b173e5a4e21e3d6147a2198876dcaebb59f3e0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-06 07:26:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a16fddf83104092109d37372b8d8dd53d083a274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:34:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/5/check/openstack-tox-pep8/262f64a/ : SUCCESS in 7m 36s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/5/check/openstack-tox-py27/3d3eb2a/ : SUCCESS in 7m 49s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/5/check/openstack-tox-py36/15bf5c2/ : SUCCESS in 7m 26s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/5/check/openstack-tox-py37/9857a32/ : SUCCESS in 7m 52s\n- openstack-tox-docs http://logs.openstack.org/05/662305/5/check/openstack-tox-docs/38d4021/html/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/662305/5/check/openstack-tox-lower-constraints/cf75eee/ : FAILURE in 6m 16s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/662305/5/check/openstack-tox-cover/c0779f5/cover/ : SUCCESS in 7m 34s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/5/check/legacy-tempest-dsvm-networking-bagpipe/59235dd/ : SUCCESS in 51m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/05/662305/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/e294ab4/ : TIMED_OUT in 2h 05m 17s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/05/662305/5/check/legacy-networking-bagpipe-dsvm-fullstack/4830af0/ : FAILURE in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5cf07dd976a9201d707e2f5bca679a9583ef90dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-06 09:44:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a73e23656afbda98f49d5dad8b9089fdbcf0f38a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:53:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/5/check/openstack-tox-pep8/04cec53/ : SUCCESS in 6m 34s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/5/check/openstack-tox-py27/4f7869b/ : SUCCESS in 6m 47s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/5/check/openstack-tox-py36/6471592/ : SUCCESS in 10m 36s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/5/check/openstack-tox-py37/4765a6f/ : SUCCESS in 8m 15s\n- openstack-tox-docs http://logs.openstack.org/05/662305/5/check/openstack-tox-docs/26c36b1/html/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/662305/5/check/openstack-tox-lower-constraints/b282296/ : FAILURE in 12m 39s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/662305/5/check/openstack-tox-cover/95bf9a3/cover/ : SUCCESS in 10m 10s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/5/check/legacy-tempest-dsvm-networking-bagpipe/685bfb9/ : SUCCESS in 53m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/05/662305/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/21fa78d/ : SUCCESS in 1h 40m 40s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/05/662305/5/check/legacy-networking-bagpipe-dsvm-fullstack/a0f7de9/ : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56d00469051520ffe42d96596c2e4830363ee5d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:53:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"78a7b77a9e5adb4b05a55ce1abfb59baa17c1bc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:37:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/662305/5/gate/openstack-tox-pep8/799c893/ : SUCCESS in 8m 57s\n- openstack-tox-py27 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py27/999fb21/ : SUCCESS in 7m 43s\n- openstack-tox-py36 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py36/1152f48/ : SUCCESS in 7m 02s\n- openstack-tox-py37 http://logs.openstack.org/05/662305/5/gate/openstack-tox-py37/f49bac5/ : SUCCESS in 7m 57s\n- openstack-tox-docs http://logs.openstack.org/05/662305/5/gate/openstack-tox-docs/d63fe9d/html/ : SUCCESS in 5m 12s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/05/662305/5/gate/legacy-tempest-dsvm-networking-bagpipe/57c5c6a/ : SUCCESS in 56m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/05/662305/5/gate/neutron-tempest-plugin-bgpvpn-bagpipe/9902b67/ : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":5},{"id":"91c77a534c34b2dd0157ac48059fc93af005d91d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:37:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6367219a6d01d588b99ac3896644497dc2130246","revisions":{"380b9449e4c2a7641204d7a89af359e5e8e9868d":{"kind":"REWORK","_number":1,"created":"2019-05-30 19:55:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/662305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/662305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/1"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:54:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:56:54.000000000","tz":120},"subject":"Switch to native openflow implementation","message":"Switch to native openflow implementation\n\nNeutron removed ovs_ofctl interface to manage openflows in [1],\nso networking_bagpipe should also stop using it and switch to\nnative implementation based on OSKen.\n\nCloses-Bug: #1831127\n\n[1] https://review.opendev.org/#/c/599496/\n\nChange-Id: I7d571386b29234c627f973bfac2005a5577e54ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/380b9449e4c2a7641204d7a89af359e5e8e9868d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/380b9449e4c2a7641204d7a89af359e5e8e9868d"}]},"branch":"refs/heads/master"},"797763909a66711ea2cd752c8205d1f81cd744ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-30 21:33:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/662305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/662305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/2"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:54:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 13:43:44.000000000","tz":120},"subject":"Switch to native openflow implementation","message":"Switch to native openflow implementation\n\nNeutron removed ovs_ofctl interface to manage openflows in [1],\nso networking_bagpipe should also stop using it and switch to\nnative implementation based on OSKen.\n\nCloses-Bug: #1831127\n\n[1] https://review.opendev.org/#/c/599496/\n\nDepends-On: https://review.opendev.org/#/c/662231/\nChange-Id: I7d571386b29234c627f973bfac2005a5577e54ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/797763909a66711ea2cd752c8205d1f81cd744ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/797763909a66711ea2cd752c8205d1f81cd744ab"}]},"branch":"refs/heads/master"},"c0ec2fecdc902e824c8429b79ff5343e0a89735f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-31 08:34:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/662305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/662305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/3"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:54:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:39:01.000000000","tz":120},"subject":"Switch to native openflow implementation","message":"Switch to native openflow implementation\n\nNeutron removed ovs_ofctl interface to manage openflows in [1],\nso networking_bagpipe should also stop using it and switch to\nnative implementation based on OSKen.\n\nCloses-Bug: #1831127\n\n[1] https://review.opendev.org/#/c/599496/\n\nDepends-On: https://review.opendev.org/#/c/662231/\nDepends-On: https://review.opendev.org/662402\n\nChange-Id: I7d571386b29234c627f973bfac2005a5577e54ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c0ec2fecdc902e824c8429b79ff5343e0a89735f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c0ec2fecdc902e824c8429b79ff5343e0a89735f"}]},"branch":"refs/heads/master"},"b4ba9b56b4a97c90d9770e2be3e1289e569cb95d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-05-31 09:12:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/662305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/662305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/4"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:54:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 02:20:15.000000000","tz":120},"subject":"Switch to native openflow implementation","message":"Switch to native openflow implementation\n\nNeutron removed ovs_ofctl interface to manage openflows in [1],\nso networking_bagpipe should also stop using it and switch to\nnative implementation based on OSKen.\n\nCloses-Bug: #1831127\n\n[1] https://review.opendev.org/#/c/599496/\n\nDepends-On: https://review.opendev.org/662231\nDepends-On: https://review.opendev.org/662402\nDepends-On: https://review.opendev.org/662142\n\nChange-Id: I7d571386b29234c627f973bfac2005a5577e54ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4ba9b56b4a97c90d9770e2be3e1289e569cb95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4ba9b56b4a97c90d9770e2be3e1289e569cb95d"}]},"branch":"refs/heads/master"},"6367219a6d01d588b99ac3896644497dc2130246":{"kind":"REWORK","_number":5,"created":"2019-05-31 09:59:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/662305/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/662305/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/662305/5"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-30 11:54:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 03:08:46.000000000","tz":120},"subject":"Switch to native openflow implementation","message":"Switch to native openflow implementation\n\nNeutron removed ovs_ofctl interface to manage openflows in [1],\nso networking_bagpipe should also stop using it and switch to\nnative implementation based on OSKen.\n\nTo fix gate issues, this commit also changes old\nlegacy-tempest-dsvm-networking-bgpvpn-bagpipe CI job to new\nneutron-tempest-plugin-bgpvpn-bagpipe which was introduced in [2].\n\nCloses-Bug: #1831127\n\n[1] https://review.opendev.org/#/c/599496/\n[2] https://review.opendev.org/#/c/657793/\n\nChange-Id: I7d571386b29234c627f973bfac2005a5577e54ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6367219a6d01d588b99ac3896644497dc2130246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6367219a6d01d588b99ac3896644497dc2130246"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
