)]}'
{"id":"openstack%2Fnetworking-bagpipe~662408","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ibf46b4bbda9b44586d236752a2bad09820dd3761","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1831127","hashtags":[],"change_id":"Ibf46b4bbda9b44586d236752a2bad09820dd3761","subject":"Update lower-constraints to valid with recent changes in Neutron","status":"MERGED","created":"2019-05-31 08:50:23.000000000","updated":"2019-06-06 17:40:57.000000000","submitted":"2019-06-06 17:40:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662408-1559842857380-95369c51","meta_rev_id":"034e0bd28ee3539da3fdb40d83570bf9bf10d4d0","_number":662408,"virtual_id_number":662408,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"date":"2019-06-06 13:58:48.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-06-06 17:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-06-05 21:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-05-31 13:55:17.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-05-31 13:56:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-05-31 13:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2019-06-05 21:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 13:55:17.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-31 13:56:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-31 13:59:24.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-06-05 21:42:33.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-06 17:40:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"369dbfe0bbffbef5227d9a88f377fd06fb86707a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 08:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8321a0d3937f872bf45eb954db4d55d629317fe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:13:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"51494d5a8fc90b08930055d01be3caaab1d70cb6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:19:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64cdef14d4f9023264e0c75524e8ea1ffa9b8cab","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:36:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f85ee41837d39e731b7e4da99c87927812fc9701","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-31 09:59:52.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9279bb14e0e8d869ec90c5ad0c44ca884c06fbd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:51:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/662408/5/check/openstack-tox-pep8/cae7862/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/08/662408/5/check/openstack-tox-py27/2a23617/ : SUCCESS in 8m 02s\n- openstack-tox-py36 http://logs.openstack.org/08/662408/5/check/openstack-tox-py36/0b36f1e/ : SUCCESS in 9m 31s\n- openstack-tox-py37 http://logs.openstack.org/08/662408/5/check/openstack-tox-py37/cfb13cc/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/08/662408/5/check/openstack-tox-docs/4140834/html/ : SUCCESS in 6m 36s\n- build-openstack-releasenotes http://logs.openstack.org/08/662408/5/check/build-openstack-releasenotes/00d659a/html/ : SUCCESS in 3m 36s\n- requirements-check http://logs.openstack.org/08/662408/5/check/requirements-check/9e782cc/ : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/662408/5/check/openstack-tox-lower-constraints/a5fe62d/ : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/08/662408/5/check/openstack-tox-cover/cd95de4/cover/ : SUCCESS in 6m 42s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/08/662408/5/check/legacy-tempest-dsvm-networking-bagpipe/7e31047/ : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/08/662408/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/d5b5d08/ : SUCCESS in 1h 45m 28s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/08/662408/5/check/legacy-networking-bagpipe-dsvm-fullstack/36265ad/ : FAILURE in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a962dfeb45dbed36b3ca32c1cf8f20116ace61a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-31 13:55:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"789b24a70ba020e89c2eacce9b8bbfc0b21811c5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-31 13:56:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e920d21f85e1328379558743009d7f16e4a7b9cd","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-31 13:59:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9a0eb19a2b347385b127611eebd127a0794bfbfc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-05 21:42:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5fd070d538ab712f33aac6bdbce34917d321ab8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:42:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"10c549d15c6c15ab2840acdeda9ce96066ca013d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:42:04.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"6a093f1dd0af99245e074abff51025eba4fe8688","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-06 13:58:48.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0ba402f3315e59e384d38da45f695f83868420ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:45:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/662408/5/check/openstack-tox-pep8/89e6893/ : SUCCESS in 8m 31s\n- openstack-tox-py27 http://logs.openstack.org/08/662408/5/check/openstack-tox-py27/edd4be3/ : SUCCESS in 7m 27s\n- openstack-tox-py36 http://logs.openstack.org/08/662408/5/check/openstack-tox-py36/9696772/ : SUCCESS in 7m 58s\n- openstack-tox-py37 http://logs.openstack.org/08/662408/5/check/openstack-tox-py37/d53181e/ : SUCCESS in 8m 10s\n- openstack-tox-docs http://logs.openstack.org/08/662408/5/check/openstack-tox-docs/8202aa7/html/ : SUCCESS in 5m 23s\n- build-openstack-releasenotes http://logs.openstack.org/08/662408/5/check/build-openstack-releasenotes/63fa539/html/ : SUCCESS in 4m 16s\n- requirements-check http://logs.openstack.org/08/662408/5/check/requirements-check/24073d5/ : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/662408/5/check/openstack-tox-lower-constraints/c7e9ba5/ : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/08/662408/5/check/openstack-tox-cover/a2e3de1/cover/ : SUCCESS in 8m 24s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/08/662408/5/check/legacy-tempest-dsvm-networking-bagpipe/d30c0db/ : SUCCESS in 49m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/08/662408/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/625ce6c/ : SUCCESS in 1h 34m 44s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/08/662408/5/check/legacy-networking-bagpipe-dsvm-fullstack/5594da8/ : FAILURE in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69812929436a26519c377a1fa0c989bb71ce3958","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:46:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1b5e6e7c993ad05d5890006e69c6ea8936cf9832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:40:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/662408/5/gate/openstack-tox-pep8/581e020/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/08/662408/5/gate/openstack-tox-py27/0c205a0/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/08/662408/5/gate/openstack-tox-py36/140e42a/ : SUCCESS in 8m 01s\n- openstack-tox-py37 http://logs.openstack.org/08/662408/5/gate/openstack-tox-py37/2c52248/ : SUCCESS in 8m 10s\n- openstack-tox-docs http://logs.openstack.org/08/662408/5/gate/openstack-tox-docs/f4c307e/html/ : SUCCESS in 5m 10s\n- build-openstack-releasenotes http://logs.openstack.org/08/662408/5/gate/build-openstack-releasenotes/ef2cabe/html/ : SUCCESS in 3m 56s\n- requirements-check http://logs.openstack.org/08/662408/5/gate/requirements-check/97d355a/ : SUCCESS in 3m 43s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/08/662408/5/gate/legacy-tempest-dsvm-networking-bagpipe/4c3ac1a/ : SUCCESS in 59m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/08/662408/5/gate/neutron-tempest-plugin-bgpvpn-bagpipe/9df3f9a/ : SUCCESS in 1h 45m 28s","accounts_in_message":[],"_revision_number":5},{"id":"802ccba5c16608fe979c3dc184cc1ebc64a8579e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:40:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"16151fcee4b5d79fd9937391dd9636fa160218c3","revisions":{"c25c975948855cd159be3cecec77ea6b01fc9a78":{"kind":"REWORK","_number":1,"created":"2019-05-31 08:50:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/662408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/662408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/1"}}},"commit":{"parents":[{"commit":"c0ec2fecdc902e824c8429b79ff5343e0a89735f","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c0ec2fecdc902e824c8429b79ff5343e0a89735f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"subject":"Update lower-constraints to valid with recent changes in Neutron","message":"Update lower-constraints to valid with recent changes in Neutron\n\nBecause of recent changes in Neutron, like remove of_interface config\noption or tranition from ryu to os-ken project, it is needed to\nbump minumum required versions of projects like neutron for\nnetworking-bagpipe.\n\nThis commit updates also versions of other libs to be the same as are\nconfigured in neutron.\n\nChange-Id: Ibf46b4bbda9b44586d236752a2bad09820dd3761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c25c975948855cd159be3cecec77ea6b01fc9a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c25c975948855cd159be3cecec77ea6b01fc9a78"}]},"branch":"refs/heads/master"},"7512bd91cae78fc045ca1fc08d54be4af13e7dfe":{"kind":"NO_CHANGE","_number":2,"created":"2019-05-31 09:13:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/662408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/662408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/2"}}},"commit":{"parents":[{"commit":"b4ba9b56b4a97c90d9770e2be3e1289e569cb95d","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4ba9b56b4a97c90d9770e2be3e1289e569cb95d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 09:13:04.000000000","tz":0},"subject":"Update lower-constraints to valid with recent changes in Neutron","message":"Update lower-constraints to valid with recent changes in Neutron\n\nBecause of recent changes in Neutron, like remove of_interface config\noption or tranition from ryu to os-ken project, it is needed to\nbump minumum required versions of projects like neutron for\nnetworking-bagpipe.\n\nThis commit updates also versions of other libs to be the same as are\nconfigured in neutron.\n\nChange-Id: Ibf46b4bbda9b44586d236752a2bad09820dd3761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7512bd91cae78fc045ca1fc08d54be4af13e7dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7512bd91cae78fc045ca1fc08d54be4af13e7dfe"}]},"branch":"refs/heads/master"},"4673fa0ec1ed3b11d495a558e16e566765a5daa6":{"kind":"REWORK","_number":3,"created":"2019-05-31 09:19:00.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/662408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/662408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/3"}}},"commit":{"parents":[{"commit":"b4ba9b56b4a97c90d9770e2be3e1289e569cb95d","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4ba9b56b4a97c90d9770e2be3e1289e569cb95d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 02:27:41.000000000","tz":120},"subject":"Update lower-constraints to valid with recent changes in Neutron","message":"Update lower-constraints to valid with recent changes in Neutron\n\nBecause of recent changes in Neutron, like remove of_interface config\noption or tranition from ryu to os-ken project, it is needed to\nbump minumum required versions of projects like neutron for\nnetworking-bagpipe.\n\nThis commit updates also versions of other libs to be the same as are\nconfigured in neutron.\n\nChange-Id: Ibf46b4bbda9b44586d236752a2bad09820dd3761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4673fa0ec1ed3b11d495a558e16e566765a5daa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4673fa0ec1ed3b11d495a558e16e566765a5daa6"}]},"branch":"refs/heads/master"},"c8c1477ca9026e8096241b8dae288b19df1bc3cf":{"kind":"REWORK","_number":4,"created":"2019-05-31 09:36:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/662408/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/662408/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/4"}}},"commit":{"parents":[{"commit":"b4ba9b56b4a97c90d9770e2be3e1289e569cb95d","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b4ba9b56b4a97c90d9770e2be3e1289e569cb95d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 02:46:00.000000000","tz":120},"subject":"Update lower-constraints to valid with recent changes in Neutron","message":"Update lower-constraints to valid with recent changes in Neutron\n\nBecause of recent changes in Neutron, like remove of_interface config\noption or tranition from ryu to os-ken project, it is needed to\nbump minumum required versions of projects like neutron for\nnetworking-bagpipe.\n\nThis commit updates also versions of other libs to be the same as are\nconfigured in neutron.\n\nThis commit updates also sphinx requirements to be the same as for\nother OpenStack projects.\n\nChange-Id: Ibf46b4bbda9b44586d236752a2bad09820dd3761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8c1477ca9026e8096241b8dae288b19df1bc3cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8c1477ca9026e8096241b8dae288b19df1bc3cf"}]},"branch":"refs/heads/master"},"16151fcee4b5d79fd9937391dd9636fa160218c3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-31 09:59:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/662408/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/662408/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/662408/5"}}},"commit":{"parents":[{"commit":"6367219a6d01d588b99ac3896644497dc2130246","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6367219a6d01d588b99ac3896644497dc2130246"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 01:53:48.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-31 09:59:52.000000000","tz":0},"subject":"Update lower-constraints to valid with recent changes in Neutron","message":"Update lower-constraints to valid with recent changes in Neutron\n\nBecause of recent changes in Neutron, like remove of_interface config\noption or tranition from ryu to os-ken project, it is needed to\nbump minumum required versions of projects like neutron for\nnetworking-bagpipe.\n\nThis commit updates also versions of other libs to be the same as are\nconfigured in neutron.\n\nThis commit updates also sphinx requirements to be the same as for\nother OpenStack projects.\n\nChange-Id: Ibf46b4bbda9b44586d236752a2bad09820dd3761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/16151fcee4b5d79fd9937391dd9636fa160218c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/16151fcee4b5d79fd9937391dd9636fa160218c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
