)]}'
{"id":"openstack%2Fnetworking-bagpipe~664672","triplet_id":"openstack%2Fnetworking-bagpipe~master~I21b7d6311383eb6cf0b696295c0b2d152ac5d054","project":"openstack/networking-bagpipe","branch":"master","topic":"bp/neutronlib-decouple-db","hashtags":[],"change_id":"I21b7d6311383eb6cf0b696295c0b2d152ac5d054","subject":"use object common_types from neutron-lib","status":"MERGED","created":"2019-06-11 19:02:59.000000000","updated":"2019-12-03 17:52:27.000000000","submitted":"2019-12-03 17:49:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664672-1575395386324-8403a22a","meta_rev_id":"37300e42805167d912363427cd1d32f30cff3b7a","_number":664672,"virtual_id_number":664672,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 17:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-11-28 09:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-11-28 10:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-11-22 13:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-11-22 13:03:47.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-11-22 13:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-11-28 21:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 14:28:32.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-07-16 14:28:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-22 12:28:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-11-28 09:09:44.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-11-28 10:31:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-11-28 21:35:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-03 17:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e3d87de7f7425e78f7b08941e9917a4785f9b01","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-11 19:02:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7744121388c8a3aef7cfd2ebb7e46001185bda14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:01:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/664672/1/check/openstack-tox-pep8/89ed27f/ : SUCCESS in 7m 18s\n- openstack-tox-py27 http://logs.openstack.org/72/664672/1/check/openstack-tox-py27/514764b/ : FAILURE in 7m 19s\n- openstack-tox-py36 http://logs.openstack.org/72/664672/1/check/openstack-tox-py36/d3e43a1/ : FAILURE in 8m 24s\n- openstack-tox-py37 http://logs.openstack.org/72/664672/1/check/openstack-tox-py37/fb56a22/ : FAILURE in 9m 15s\n- openstack-tox-docs http://logs.openstack.org/72/664672/1/check/openstack-tox-docs/042a171/html/ : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/664672/1/check/openstack-tox-lower-constraints/aed0e12/ : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/72/664672/1/check/openstack-tox-cover/2a08ee6/ : FAILURE in 7m 35s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/72/664672/1/check/legacy-tempest-dsvm-networking-bagpipe/f5f0707/ : SUCCESS in 59m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/72/664672/1/check/neutron-tempest-plugin-bgpvpn-bagpipe/3c45bea/ : SUCCESS in 1h 37m 42s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/72/664672/1/check/legacy-networking-bagpipe-dsvm-fullstack/a4078a9/ : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9236bb3b2475822d2f862146825454647aba3463","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-08 20:13:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"709596b800bf88168672d3b62c9571e124dc4302","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 21:39:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/664672/2/check/openstack-tox-pep8/74e7ee0/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/72/664672/2/check/openstack-tox-py27/6c3c613/ : SUCCESS in 6m 41s\n- openstack-tox-py36 http://logs.openstack.org/72/664672/2/check/openstack-tox-py36/5f7d88e/ : SUCCESS in 6m 18s\n- openstack-tox-py37 http://logs.openstack.org/72/664672/2/check/openstack-tox-py37/292a373/ : SUCCESS in 6m 35s\n- openstack-tox-docs http://logs.openstack.org/72/664672/2/check/openstack-tox-docs/3f617db/html/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/664672/2/check/openstack-tox-lower-constraints/bfc8841/ : FAILURE in 6m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/72/664672/2/check/openstack-tox-cover/582883f/cover/ : SUCCESS in 7m 06s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/72/664672/2/check/legacy-tempest-dsvm-networking-bagpipe/2afc235/ : SUCCESS in 1h 01m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/72/664672/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/f51c01e/ : SUCCESS in 1h 23m 49s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/72/664672/2/check/legacy-networking-bagpipe-dsvm-fullstack/e6cda05/ : FAILURE in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c89d1a7231fd8d0dd840c2a282f965f8be20920d","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-16 14:28:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16d6591f8d55bb2c251ecfa35ae8c8dd15acfcfd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-16 14:28:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"076f5c52639707b1519db05c73c5abaacd988fa8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-16 14:28:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7c09dc2f6581077764122306c82da58ce1d2fbe5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-22 12:28:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4081a5635bbb00fc5d6cd8ed4e5abd3fae518b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 12:28:24.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"654fe8e26a15c859e9a6be1efb7a540ba7a2a458","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-22 13:03:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4e5a4185fd591d8bedb4ed3da0fbb0881684cff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:04:04.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"1480019fc0b401b547e299a4cbfff4ffedb33423","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-27 15:43:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"90ef0bd3ef4c018835bf4b19dea1d413491bba39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:27:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3481cbb3c2349de875d47687f941c28 : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc9618d3f0d0486da6e390a09fa3ca94 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf920356489f46f68f4dbce80d9f4a76 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0f08dda286e48cf86b5d6ee46792935 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77899dd87474db1839fa029fc79a060 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5d2571151748ed8490dd784ad25a84 : FAILURE in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab7e79821ff34e61890eb7bd7cca5164 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/546ad53817ec4c4fad8fc6ce798bb8a3 : SUCCESS in 1h 28m 56s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f866ff38cf5448a79c712d9dca2f8614 : FAILURE in 17m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ef4167aeb61b444caf62bf108a1ced76 : SUCCESS in 1h 36m 41s","accounts_in_message":[],"_revision_number":3},{"id":"6f066b49511e0a48c431a2a6777a39a643114c7b","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-28 09:09:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"42776f919f6c7c71016a0f4fcc5ad812934e3df3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-28 10:31:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2d3322cd05b6700171d7f483d5c08bca7f2ed4f1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-28 21:35:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"85f17f6fd2497cc18a09de625b01ee706ecc2736","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-03 14:51:14.000000000","message":"Patch Set 3:\n\nrecheck neutron patch is merged","accounts_in_message":[],"_revision_number":3},{"id":"ee9bf86f0538e285dae4f701ae188e4780f2ac99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:25:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe89403bbae427daa9572d7a7fcfa96 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3184b3c1cea848b58b58ebea0e11297f : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c35de19ba2542108b1be7d50f739c62 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/855881c8006841068826ff1bb99938d5 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f37fb247ca430ca36b48cb45f39211 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2666b4f07fe2486bb55d575cac13f6fb : FAILURE in 7m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132412eb2f984b23a35d66dc96ed6762 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c80bae752e9644f981d1d17c8e0660ab : SUCCESS in 1h 28m 18s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/869d3630286747d1b9220f75514a19a6 : FAILURE in 18m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2552d10709f1442185f00b155d7bcf58 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":3},{"id":"21e1662e406798e6aaf5fc41b88d9c0dbd81d3f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:25:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a6d460cda9b2c5fad0b8583f8bb019d76612c4d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:49:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed37fe179f74d57b20f2821478ea6b2 : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b6944dc95934924877d29449e4c92a0 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5e676e51c8d410ab522c6008beefeee : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6d23dc7c7d64f10b7aa0b4f83f21cbf : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d274399cb04ddab2f01ca4b106f6f4 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/47e3f52c6f0d4d76a22f8f330990a12f : SUCCESS in 1h 22m 51s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0b5d08dba88740299845e5c7d323861d : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":3},{"id":"eb42c9f08f60637728dff33ff64840cfa68ccfd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:49:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e8cfe069038f283a19b56530603abdcaa15ca115","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:52:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ac3d067f6a4c6383cacae62a8ea761 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"410cbb01c1a6d7d392b1a1b322b14126aaf2be29","revisions":{"a9d9320b147760dd6d2d73c9ea7fdd0b68f75a8e":{"kind":"REWORK","_number":1,"created":"2019-06-11 19:02:59.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/72/664672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/664672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/1"}}},"commit":{"parents":[{"commit":"16151fcee4b5d79fd9937391dd9636fa160218c3","subject":"Update lower-constraints to valid with recent changes in Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/16151fcee4b5d79fd9937391dd9636fa160218c3"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-11 19:02:54.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-11 19:02:54.000000000","tz":-360},"subject":"use object common_types from neutron-lib","message":"use object common_types from neutron-lib\n\nSwitch over to neutron-lib for OVO common_types.\n\nChange-Id: I21b7d6311383eb6cf0b696295c0b2d152ac5d054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9d9320b147760dd6d2d73c9ea7fdd0b68f75a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9d9320b147760dd6d2d73c9ea7fdd0b68f75a8e"}]},"branch":"refs/heads/master"},"eb1bf62bed6602a84551386535079bd9ee7b589e":{"kind":"REWORK","_number":2,"created":"2019-07-08 20:13:15.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/72/664672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/664672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/2"}}},"commit":{"parents":[{"commit":"8f19c35b5ead980bcf2aa479eab0703425f6ee80","subject":"Merge \"Add local bindep.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8f19c35b5ead980bcf2aa479eab0703425f6ee80"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-11 19:02:54.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-08 20:12:59.000000000","tz":-360},"subject":"use object common_types from neutron-lib","message":"use object common_types from neutron-lib\n\nSwitch over to neutron-lib for OVO common_types.\n\nDepends-On: https://review.opendev.org/#/c/664646/\n\nChange-Id: I21b7d6311383eb6cf0b696295c0b2d152ac5d054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/eb1bf62bed6602a84551386535079bd9ee7b589e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/eb1bf62bed6602a84551386535079bd9ee7b589e"}]},"branch":"refs/heads/master"},"410cbb01c1a6d7d392b1a1b322b14126aaf2be29":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-22 13:03:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/664672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/664672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/664672/3"}}},"commit":{"parents":[{"commit":"f4113085b069683f851e475c8bc738ce4d7478b2","subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4113085b069683f851e475c8bc738ce4d7478b2"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-11 19:02:54.000000000","tz":-360},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2019-11-22 13:03:47.000000000","tz":0},"subject":"use object common_types from neutron-lib","message":"use object common_types from neutron-lib\n\nSwitch over to neutron-lib for OVO common_types.\n\nDepends-On: https://review.opendev.org/#/c/664646/\n\nChange-Id: I21b7d6311383eb6cf0b696295c0b2d152ac5d054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/410cbb01c1a6d7d392b1a1b322b14126aaf2be29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/410cbb01c1a6d7d392b1a1b322b14126aaf2be29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
